)]}'
{"id":"openstack%2Fpuppet-mistral~714374","triplet_id":"openstack%2Fpuppet-mistral~master~I06a4722484f17a9027840400d2a522a7d271bfba","project":"openstack/puppet-mistral","branch":"master","topic":"remove_idle_timeout_option","hashtags":[],"change_id":"I06a4722484f17a9027840400d2a522a7d271bfba","subject":"Remove idle_timeout option","status":"MERGED","created":"2020-03-23 06:47:15.000000000","updated":"2020-03-24 01:06:16.000000000","submitted":"2020-03-24 01:05:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714374-1585011900799-046e596a","meta_rev_id":"98b249fa6a005de1b5b070330c2765fb55355cdd","_number":714374,"virtual_id_number":714374,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 01:05:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-23 13:01:15.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-23 23:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-23 23:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 13:01:15.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-23 23:42:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-24 01:05:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9c566ac777923f9ef33b244a9cc6aae0e9cad87","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-23 06:47:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76b8e2d62ce8309fabc1fabb2ba97c1f5feea86b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 08:11:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f1e0d44ebcd47feb2454a4cd55776b0 : SUCCESS in 6m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e16803020f847ab90d5ddcea6e5a43d : SUCCESS in 6m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/112d0b9a35254ffa892fc079cfed7154 : SUCCESS in 8m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b692f75c04204049a7561d9ed8a6b6c3 : SUCCESS in 15m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e4a090a4533d4d159935b06b0b9bc95f : SUCCESS in 12m 28s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/27bab2d6a26348a4ab9e6b8b21d12e44 : SUCCESS in 12m 37s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b3125f5452741d3a564d004acfcba27 : SUCCESS in 12m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecbc8ba3bb954934b5c22e14ed76b5c8 : SUCCESS in 10m 33s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f18fc2019a04c96aa184336b9ff71ce : SUCCESS in 36m 09s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/710c9851587842219a366939cae63a0c : SUCCESS in 1h 15m 20s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eb29cd1366e64300af207ee9649a032f : SUCCESS in 1h 01m 31s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/26836dbbe199415bb71b107fd5cc728b : SUCCESS in 21m 17s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8804adea97c3414996dffe1772b8d4c9 : FAILURE in 9m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d00c41f5663a4f40994f61d410982ac7 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"92860d49432704bad717f49d043ab5fb4b53f049","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-23 13:01:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c8f894ee953c18efc76998395c9982240346068","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-23 23:42:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"848d046969b3211be5e36a7b8e32d53ab5d32e55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 23:42:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03bad1ed10842b9c32795ce17683842316a4d73b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:05:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff855611188004d44f8f91bfb9a1acbddee80f5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:05:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9f47f26eda2498285db9f88b9909b02 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/087057aa42474444b91a16fd9dece271 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/252df0adb61f40a2946546471f09c8f4 : SUCCESS in 5m 30s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4fabf899f4cf4540bdcc5ee835f7af9d : SUCCESS in 17m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/493f33cf7c10485fa0938038be1a5b7c : SUCCESS in 9m 49s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cd994b6dd3834412b591af709e9ed0bb : SUCCESS in 1h 17m 44s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/79fdd25a13ea4d48b07f87a1c04cea78 : SUCCESS in 17m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/149534f383cd40b6bf0dcc1fd025b358 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1943a328f543c812ecf9734df9eada0ccbeccc58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 01:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c244a44592094f30b21f63315b034639 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2f0de297b3335d51d1c98df219a2ddc658b71a5","revisions":{"b2f0de297b3335d51d1c98df219a2ddc658b71a5":{"kind":"REWORK","_number":1,"created":"2020-03-23 06:47:15.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/74/714374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/74/714374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/74/714374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/74/714374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/74/714374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/74/714374/1"}}},"commit":{"parents":[{"commit":"a6f247c60924fa4ef5d9d5aa3754d9106ffa2629","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/a6f247c60924fa4ef5d9d5aa3754d9106ffa2629"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-23 06:45:48.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-23 06:45:48.000000000","tz":480},"subject":"Remove idle_timeout option","message":"Remove idle_timeout option\n\nThe idle_timeout parameter has been deprecated for two releases.\nWe can remove it.\n\nChange-Id: I06a4722484f17a9027840400d2a522a7d271bfba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b2f0de297b3335d51d1c98df219a2ddc658b71a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b2f0de297b3335d51d1c98df219a2ddc658b71a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
