)]}'
{"id":"openstack%2Fpuppet-mistral~714866","triplet_id":"openstack%2Fpuppet-mistral~master~Ia6d09856e71e42bfa12b91a49735798bfed9818f","project":"openstack/puppet-mistral","branch":"master","topic":"bug/1868511","hashtags":[],"change_id":"Ia6d09856e71e42bfa12b91a49735798bfed9818f","subject":"Deprecate min_pool_size option","status":"MERGED","created":"2020-03-25 08:23:18.000000000","updated":"2020-03-25 20:10:42.000000000","submitted":"2020-03-25 20:07:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714866-1585166861688-81340386","meta_rev_id":"f028b814ebb6dd8ce771aeca77d20ee9ded54546","_number":714866,"virtual_id_number":714866,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 20:07:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2020-03-25 17:30:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-03-25 18:59:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 17:30:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-25 18:59:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-25 20:07:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e45935946ffe0071ec8176292f16b28caa198039","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-25 08:23:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d94e7013183bba8a87cf25d3eb44641f0ddad23e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 09:31:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f3a6763d7514db6be8a49a4398342be : SUCCESS in 6m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d4ac2cabb7d4d8599325150b8c5faea : SUCCESS in 5m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ce595011080422a87df7f4d38a97f78 : SUCCESS in 7m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/eab8dfc5bdc549b3a4e3dba08048aac6 : SUCCESS in 21m 09s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5d751d6c736d4e0fa32e2715ec448560 : SUCCESS in 15m 16s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/65eeca97c8b9486d91bead587bb43a0f : SUCCESS in 15m 18s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e66005c364c403ca51728157e4431a4 : SUCCESS in 13m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b0fa87ea47843dca39cdfd43078f938 : SUCCESS in 13m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/291cb160fba5437e8290b08ccdef9ad1 : SUCCESS in 38m 55s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cdf09dbe33704f93b465420c9c651ebb : SUCCESS in 1h 04m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3d93aba84d294591b667349d0b97eabc : SUCCESS in 19m 32s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a52fefc7ce324599aca8ddaf6a7cfd1a : FAILURE in 9m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7a98ecc4bf74f7f866683064e7ba8f0 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b207c491751c6f6994672bce16d2c380222e8e59","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-25 17:30:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"207ac10af5ff7c58b8f19a3de5af350f191fc983","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-25 18:59:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95d1e4ae613ef13fba96edafca91459342dde05c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 18:59:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3742a31121315f808e7b0e96ec29198f38b86e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 20:07:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a1a61433db04095b89763a34f46c20d : SUCCESS in 8m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65f27a79ed6a43fca491010908a4b0f2 : SUCCESS in 5m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5be1b1c9a4e6444a9101ec212899826b : SUCCESS in 5m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1815662b5d5a4d9ca6b47d7da52669cb : SUCCESS in 23m 51s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40a8daeaeb9b4bd1b4d68c2b39dbd1ba : SUCCESS in 11m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e9285b0fa69340e0afb3cea20bf322d3 : SUCCESS in 1h 02m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ece67490d3314818a86c7e72784c5795 : SUCCESS in 18m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2c3db76879f4b92a06637e35f38b4a0 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3bf2c88c3532dddbb2d33ea1c48823e39dee4932","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 20:07:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0e37ea7b6287ed4dcabcf13a46f456414a5fd087","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 20:10:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b97123a4f64664a38ab17682b58ec6 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a89c5a0024681dd4908fb44e576e5594949cc6d","revisions":{"5a89c5a0024681dd4908fb44e576e5594949cc6d":{"kind":"REWORK","_number":1,"created":"2020-03-25 08:23:18.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/66/714866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/66/714866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/66/714866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/66/714866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/66/714866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/66/714866/1"}}},"commit":{"parents":[{"commit":"b2f0de297b3335d51d1c98df219a2ddc658b71a5","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b2f0de297b3335d51d1c98df219a2ddc658b71a5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-25 06:53:41.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-25 06:53:41.000000000","tz":480},"subject":"Deprecate min_pool_size option","message":"Deprecate min_pool_size option\n\nmin_pool_size option is not used，see:\n\nhttps://review.opendev.org/#/c/565090/\n\nChange-Id: Ia6d09856e71e42bfa12b91a49735798bfed9818f\nCloses-Bug: #1868511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5a89c5a0024681dd4908fb44e576e5594949cc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5a89c5a0024681dd4908fb44e576e5594949cc6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
