)]}'
{"id":"openstack%2Fpuppet-mistral~715854","triplet_id":"openstack%2Fpuppet-mistral~master~Iea4d05ea795cd77563ae9a00a50460f8406710e7","project":"openstack/puppet-mistral","branch":"master","hashtags":[],"change_id":"Iea4d05ea795cd77563ae9a00a50460f8406710e7","subject":"Update mysql module to \u003c11.0.0","status":"MERGED","created":"2020-03-30 08:03:51.000000000","updated":"2020-03-31 00:51:55.000000000","submitted":"2020-03-31 00:51:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715854-1585615915912-28344ecc","meta_rev_id":"f8d26668b2ea68edc2da777b069c880e492b89fa","_number":715854,"virtual_id_number":715854,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 00:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-30 13:55:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-31 00:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-31 00:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:55:40.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-31 00:42:00.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-03-31 00:51:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95b589979840efd566b7f0ff7cde56c28f9a4964","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 08:03:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fd8792742ff1460affac35ce221bde03412432a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 08:13:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f4da6178459a4775af34881fc749b6a1 : SUCCESS in 2m 44s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72303bcf98b94711958944e5b357478d : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e4bdb88b374a09ec522c850681c6cd9907cbbced","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-30 13:55:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d2b6364b0e851df542fe7b2af7be2edff84c67e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-31 00:42:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d6b02b394d9ce4117a7f5157dfc846eddc8c9cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 00:42:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"071fa2bbb9553ba805be5d604a7afdc0a11fb457","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 00:51:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f8d26668b2ea68edc2da777b069c880e492b89fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 00:51:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/1d000faf8256465dafbcd7977b33d77e : SUCCESS in 2m 35s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c13f8d6af7b843b786942c63e9761f9a : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d381e726898eeb7890ba0b296af6f7c33428f745","revisions":{"d381e726898eeb7890ba0b296af6f7c33428f745":{"kind":"REWORK","_number":1,"created":"2020-03-30 08:03:51.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/54/715854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/54/715854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/54/715854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/54/715854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/54/715854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/54/715854/1"}}},"commit":{"parents":[{"commit":"98907df9102dc5c00789a7cda05874dd00d4fd03","subject":"Remove leftover param comments in init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/98907df9102dc5c00789a7cda05874dd00d4fd03"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 08:03:44.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 08:03:44.000000000","tz":120},"subject":"Update mysql module to \u003c11.0.0","message":"Update mysql module to \u003c11.0.0\n\nSame as was changed in openstacklib.\n\nChange-Id: Iea4d05ea795cd77563ae9a00a50460f8406710e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/d381e726898eeb7890ba0b296af6f7c33428f745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/d381e726898eeb7890ba0b296af6f7c33428f745"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
