)]}'
{"id":"openstack%2Fpuppet-mistral~716190","triplet_id":"openstack%2Fpuppet-mistral~master~I5a8b41e5a8281f7f8643c4bdbd8c38cd62df3688","project":"openstack/puppet-mistral","branch":"master","topic":"centos8-client","hashtags":[],"change_id":"I5a8b41e5a8281f7f8643c4bdbd8c38cd62df3688","subject":"Expect python3 client package in CentOS8","status":"MERGED","created":"2020-03-31 07:46:37.000000000","updated":"2020-03-31 22:05:12.000000000","submitted":"2020-03-31 22:05:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716190-1585692312495-3ad24aef","meta_rev_id":"664e8be654c437737e005bc57de9052c899b7dde","_number":716190,"virtual_id_number":716190,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 22:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-31 14:47:53.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-31 21:49:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 14:47:53.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-31 21:49:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-31 22:05:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3aca4077d272edbc17be6577f32be5991292e3e8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 07:46:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3152d7de9b6f38f68e5b9b32fa170e367c7b974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 08:08:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7245c5fd24f447a6b75371a45a52db39 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/321c542fbd0144f19439286bf0b25cf1 : SUCCESS in 6m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6754e985e745472c8614e86707cc7ec1 : SUCCESS in 5m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f43290e14d354dcba424b68dfd63dba8 : FAILURE in 17m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3657c7108c8b432291fcc43d460cbcad : FAILURE in 12m 29s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f72d87e75cb442b1a8158fcdf552bd1f : FAILURE in 14m 26s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f41ed47283844cbaf402b8f0582ffc9 : FAILURE in 13m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/724b6b9380514a08925a8d8f73c2c078 : FAILURE in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1427b6394df955c3afc42d89d93522604f7a0133","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-31 08:15:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ba5c91bacfd492b8c5f5da49f1df0bb5ef379f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 09:02:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3f0ebf4f8974844aa74feb6d4c01573 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a9f9686f62a4021b3abfc5612f021c9 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba2b949bafe0445cadd821326984345e : SUCCESS in 4m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/06817d5846494d4f8de1954f55c40c44 : SUCCESS in 21m 33s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/503aa84a40754e7cabbc89d151bf443e : SUCCESS in 11m 41s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d376bd760c08428799c0e9ab4d198e4d : SUCCESS in 11m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/585aff19b5914a9fb9f973807993bfe3 : SUCCESS in 13m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/729d006acc744bd9b8b13007cf015a90 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2a99f0caf3bb1b9db56837fbcd67a0100da8eab","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-31 14:47:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1e04cf7e4793df03083112ed9926c60108ffb30c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-31 21:49:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c75f317ec1d3f349abebf76a5a98bf1738570dee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 21:49:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d9b6632d3a8c0282b652380a5f76d91844bcbef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 22:05:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6660406d02642dca1863c47431b6277 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfd15182e3594f97ae73ff4f697a2a17 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4649962e0564427b75116382f03b66b : SUCCESS in 5m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6ae4ac31aa4d4d18a5b7b2cdc618e1e7 : SUCCESS in 12m 47s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4e7d9aefc8704b5f874ba002b161d458 : SUCCESS in 12m 50s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/34564129d13043909b43c4abbc395984 : SUCCESS in 12m 39s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3862b4857fc7425aa9f07bca0d7e9cb1 : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":2},{"id":"511bdd1b4ac76758ba384e6c060e1616fed4e40f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 22:05:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"947da56b58c3afc1f24ccbf720a068de5173c842","revisions":{"23b486440bae7fc58674bf9a10c639d39f201344":{"kind":"REWORK","_number":1,"created":"2020-03-31 07:46:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/716190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/90/716190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/716190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/716190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/716190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/90/716190/1"}}},"commit":{"parents":[{"commit":"5a89c5a0024681dd4908fb44e576e5594949cc6d","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5a89c5a0024681dd4908fb44e576e5594949cc6d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:45:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:46:34.000000000","tz":540},"subject":"Expect python3 client package in CentOS8","message":"Expect python3 client package in CentOS8\n\nIn CentOS, we expect to have python3 client package in 8.x while we\nexpect to have python2 in 7.x .\nFix unit tests to expect the correct version according to os major\nversion.\n\nChange-Id: I5a8b41e5a8281f7f8643c4bdbd8c38cd62df3688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/23b486440bae7fc58674bf9a10c639d39f201344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/23b486440bae7fc58674bf9a10c639d39f201344"}]},"branch":"refs/heads/master"},"947da56b58c3afc1f24ccbf720a068de5173c842":{"kind":"REWORK","_number":2,"created":"2020-03-31 08:15:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/716190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/90/716190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/716190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/716190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/716190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/90/716190/2"}}},"commit":{"parents":[{"commit":"5a89c5a0024681dd4908fb44e576e5594949cc6d","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5a89c5a0024681dd4908fb44e576e5594949cc6d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 07:45:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-31 08:15:15.000000000","tz":540},"subject":"Expect python3 client package in CentOS8","message":"Expect python3 client package in CentOS8\n\nIn CentOS, we expect to have python3 client package in 8.x while we\nexpect to have python2 in 7.x .\nFix unit tests to expect the correct version according to os major\nversion.\n\nChange-Id: I5a8b41e5a8281f7f8643c4bdbd8c38cd62df3688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/947da56b58c3afc1f24ccbf720a068de5173c842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/947da56b58c3afc1f24ccbf720a068de5173c842"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
