)]}'
{"id":"openstack%2Fpuppet-mistral~729219","triplet_id":"openstack%2Fpuppet-mistral~master~I6b4f8479030b28957c540c5f0775d5f799dff5dd","project":"openstack/puppet-mistral","branch":"master","topic":"db-password","hashtags":[],"change_id":"I6b4f8479030b28957c540c5f0775d5f799dff5dd","subject":"Remove password hash generation in each puppet modules","status":"MERGED","created":"2020-05-19 11:21:55.000000000","updated":"2020-05-20 04:43:21.000000000","submitted":"2020-05-20 04:43:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729219-1589949801829-cbaaa819","meta_rev_id":"f434d5632420f4cedd4e33813099a88fc5065efc","_number":729219,"virtual_id_number":729219,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-20 02:34:52.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 04:43:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-19 12:12:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-05-19 21:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-05-19 21:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 12:12:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-19 21:47:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-20 04:43:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c95c0d5d763d9913be0357e4be496a9abeb1bd84","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 11:21:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2926610cba7e1dc443055c95e1c55d40528308a7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-19 12:12:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c6618614f5837270145a8cd6b6b17a1a65c0b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:31:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/30e9d5cee16e4c0493a0f830385c2239 : SUCCESS in 2m 51s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a86eacab8584389b68bd30b34104089 : SUCCESS in 5m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be26de51dd034b98bdaddb5df87f3e87 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16b3960e9e904002ad01f1f8e432e891 : SUCCESS in 5m 37s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c17b3a69a1144768bba920b69057525b : SUCCESS in 15m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1676e93543654adaa000be6f328749f2 : SUCCESS in 18m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba061bf27e044994b9ae72d33084a738 : SUCCESS in 10m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f58e070fa4cd4fb783f5f6154d146cc1 : SUCCESS in 13m 04s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/168339cc8b2c40c794b25d2f568fdb5e : FAILURE in 39m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e417898d273c4cdaaa1655efaafb5f1f : SUCCESS in 1h 08m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/23daff806a8340eabafb80242834289d : SUCCESS in 24m 39s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/8266ad28101f4e2d8dde2ef5ccbf3eca : FAILURE in 8m 01s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da212302c94f48108e9ff9006ad0a2f7 : FAILURE in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1e14c1ab72d5878e322cb58838eb0871e2967bf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-19 21:47:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b28a9ab32f5922ac4db4668dde471233087d4f6b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-20 02:34:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23eb1efa98592eb8bed120b65dd1aaca3a37c45b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 03:44:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2a5d0a0b248f4e9ea105cbe82094a92a : SUCCESS in 3m 02s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/450f3752a76a46c895257dadb665a944 : SUCCESS in 8m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61d0a3997cc246bba836d00a495a148c : SUCCESS in 7m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f811bf1b6b064fefaf60031da408cc41 : SUCCESS in 5m 58s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/993333b517b74d74aafc8f4b0b4142dd : SUCCESS in 12m 49s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0c1a173a1e794f51a27cb16d9cd59f3f : SUCCESS in 13m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dd576caaec34be49dd9b73d41a09dc9 : SUCCESS in 11m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1cd1f81e5694de6ab9e22d1443c5d69 : SUCCESS in 11m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38c4637d331d4a398f547c24eb8b2c68 : FAILURE in 33m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/95a8906c8ae84c418241ac15e010f564 : SUCCESS in 1h 03m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fa2dd02d382b400782f68b0911f83171 : SUCCESS in 20m 50s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2053f6b1806044748d586e92a6e4312d : FAILURE in 12m 04s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d862c716671b40b18e399b58a6e7a4e0 : FAILURE in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af896d11e9b4f628bbd8761ea937ea5ab7c96b93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 03:44:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5618e024bce9897e86e70ec251380003ed2293c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 04:43:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9fd2cd84529faed64c8966b4a76e8dcd08f6d0a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 04:43:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/a4f8ea74ddaa4a3b920e058723d12c62 : SUCCESS in 3m 15s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bacec88dad946679285266f2eb9025f : SUCCESS in 4m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52b1eec4cd404e97bd886caf5515ca6b : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a9d58eba91d4e88a663a6f1d72db582 : SUCCESS in 5m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b24b560bc0dc41148c5649f490d7a203 : SUCCESS in 11m 48s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1af4ea74570843b0ba65ec0f587034f8 : SUCCESS in 14m 39s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5853831f099c4fafa0884c4a3213ab4d : SUCCESS in 9m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce3ecb1b90844152976bcb7ee2d9de97 : SUCCESS in 57m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4cedd50c52c240f09b44dea75f606fdc : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e7d912fdee63349b8fe9f357cdf26b4570e219d","revisions":{"4e7d912fdee63349b8fe9f357cdf26b4570e219d":{"kind":"REWORK","_number":1,"created":"2020-05-19 11:21:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/729219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/19/729219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/19/729219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/19/729219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/19/729219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/19/729219/1"}}},"commit":{"parents":[{"commit":"a5a8b53078808d78414b2ceb100a751022fbe450","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/a5a8b53078808d78414b2ceb100a751022fbe450"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 11:21:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 11:21:41.000000000","tz":540},"subject":"Remove password hash generation in each puppet modules","message":"Remove password hash generation in each puppet modules\n\n... and migrate it to openstacklib so that all logics about database\nconfiguration are implemented in one common place.\n\nDepends-on: https://review.opendev.org/#/c/728595/\nChange-Id: I6b4f8479030b28957c540c5f0775d5f799dff5dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/4e7d912fdee63349b8fe9f357cdf26b4570e219d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/4e7d912fdee63349b8fe9f357cdf26b4570e219d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
