)]}'
{"id":"openstack%2Fpuppet-mistral~745685","triplet_id":"openstack%2Fpuppet-mistral~master~I574c8233e15240074e67520b30df9cca79a9c20b","project":"openstack/puppet-mistral","branch":"master","hashtags":[],"change_id":"I574c8233e15240074e67520b30df9cca79a9c20b","subject":"Expose vhost_custom_fragment","status":"MERGED","created":"2020-08-11 14:43:12.000000000","updated":"2020-08-12 22:45:22.000000000","submitted":"2020-08-12 22:45:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745685-1597272322562-ff0595dd","meta_rev_id":"53e42d81ec247105012fd12ada124be251ad45a8","_number":745685,"virtual_id_number":745685,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-12 22:45:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-12 15:26:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-12 21:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-12 21:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 15:26:24.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-12 21:22:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-12 22:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a80f9f748fa85aeb3ffd947aee54a4e9dd851d77","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-11 14:43:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb5859729c56b52d4056aaf406885359da18071e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:28:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2ea5070d1c748e3830318ce4ab2232d : SUCCESS in 4m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6375003e71bf49f487b84bd823c93ac7 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/934a8c73096845edbc5e0d78808d8b28 : SUCCESS in 5m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/837749590039415ab7a22ce849e81b7b : FAILURE in 12m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8bf446bbca5647d19c7e3f2c58180567 : FAILURE in 12m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e68c99830bfb4111bd08f1f4b887e226 : FAILURE in 11m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c4b7578596244b08bdf12391d8c8f0b : FAILURE in 6m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/695b4ffa71df4cb29ff491b9f4ed4bb1 : SUCCESS in 47m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2d0528d2278649f88399fa117f4f7878 : SUCCESS in 1h 38m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/837b4b593bf44096a1f0765e37274c11 : SUCCESS in 20m 46s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/947e10aca7934521b02a775492aa0058 : FAILURE in 8m 58s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8df3fc26538442a3b03a37f97d70b085 : FAILURE in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"479a418a36eaabff37ed709cba8710a6d4d19def","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-11 21:51:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f47344d231d30dad1d03bc9dac1715bd52c6a27d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 23:26:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/233840399e994bdfacc2446891dfb195 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c7906198da54664a6348315c1aa1bfc : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51d2459723fd45ed80f19222b9177410 : SUCCESS in 3m 50s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5724ebf0cec046279581bad381b24a34 : FAILURE in 13m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d03c7f34f28846a38a922b048cd9c1a7 : FAILURE in 11m 12s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c662dbdf7d64b16a121a0630de8189d : FAILURE in 8m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/766c7791176144089bac7377afc3a90d : FAILURE in 6m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f113c18281b4296b8e0d91e6a21c5c9 : SUCCESS in 47m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b6e521b261ed4d97b85ee03d77118b14 : SUCCESS in 1h 29m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/820028f867c048edaceeb24be956b12d : SUCCESS in 13m 55s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/650c332382414c09ad844177b04242e9 : FAILURE in 8m 03s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d241bc2310394609ac2347233219866e : FAILURE in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc696fef46a53ae672a8ad8cf139a5e85c66d299","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-12 13:14:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dff0f7f48e8b5e2960131c2d77699eecf175efe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 14:49:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae08bbf36abc4b64896c2e81edb04236 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71d15def3c9e4f188c13f7480669c1bb : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed48983f4ec54f3dacd7bc2d9018137c : SUCCESS in 4m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/34bbe65edb694c2bae8403e8e3630a4a : SUCCESS in 13m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3fce8af9a8e64bc29957f854a4a59e17 : SUCCESS in 12m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/121853f558134f5782906378aa32d880 : SUCCESS in 12m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad33742df8274104b238ad719500764a : FAILURE in 8m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29849e5da7254d2cbed6ef94cc4645ea : SUCCESS in 45m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9da71fbb7e454f5c85f3a60ff3687e86 : SUCCESS in 1h 28m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cf90aa5a8d07463ab57eeb2a6946635b : SUCCESS in 19m 56s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b18af40b24704033af610ef3d5b5c054 : FAILURE in 12m 01s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0936ca7b34da41728648729a7348347d : FAILURE in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63a79f40b8dc542c15628f5ff156c4a90f20114a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-12 15:26:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6bd06082a2a89e328e6702b4123bd987005c4074","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-12 21:22:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6807205981ef3e0c39ae579427f323bfaa932e66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 21:23:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5c5110441c0d2ba784cea08b3c7722f0f1fa71c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 22:45:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"53e42d81ec247105012fd12ada124be251ad45a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 22:45:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abb6aaac6d4c42348ac7ec4d3d3218c9 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9fef5305de340e08a13e28562f9411c : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5203674dded4f82b9aa4ba885745210 : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a8827ea855a64ff2a3df36bf0be11ce8 : SUCCESS in 11m 57s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5a1740c46ac245e4b6ca86e9c0eb6ccd : SUCCESS in 14m 21s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba82ab66c2c040478117bba3c41d09ce : SUCCESS in 9m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6ef6a9f3b9834716a3a9703714dac080 : SUCCESS in 1h 20m 37s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4e58aa9fe8fe49fb9accedb2b7ace521 : SUCCESS in 36m 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"82cd962e78cab57395f61abacee6d6502067fc99","revisions":{"ec923e7f6285076c391f89b9a48e77059c6a1aa4":{"kind":"REWORK","_number":1,"created":"2020-08-11 14:43:12.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/85/745685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/85/745685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/1"}}},"commit":{"parents":[{"commit":"10f341af44e56f6dad87174da19746ca5bbccd9f","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/10f341af44e56f6dad87174da19746ca5bbccd9f"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-11 14:42:37.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-11 14:42:40.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Mistral service.\n\nChange-Id: I574c8233e15240074e67520b30df9cca79a9c20b\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/ec923e7f6285076c391f89b9a48e77059c6a1aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/ec923e7f6285076c391f89b9a48e77059c6a1aa4"}]},"branch":"refs/heads/master"},"ca0142c9748bec9a18a3e1d34134c00b21e3aead":{"kind":"REWORK","_number":2,"created":"2020-08-11 21:51:37.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/85/745685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/85/745685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/2"}}},"commit":{"parents":[{"commit":"10f341af44e56f6dad87174da19746ca5bbccd9f","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/10f341af44e56f6dad87174da19746ca5bbccd9f"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-11 14:42:37.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-11 21:51:24.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Mistral service.\n\nChange-Id: I574c8233e15240074e67520b30df9cca79a9c20b\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/ca0142c9748bec9a18a3e1d34134c00b21e3aead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/ca0142c9748bec9a18a3e1d34134c00b21e3aead"}]},"branch":"refs/heads/master"},"82cd962e78cab57395f61abacee6d6502067fc99":{"kind":"REWORK","_number":3,"created":"2020-08-12 13:14:13.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/85/745685/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/85/745685/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/85/745685/3"}}},"commit":{"parents":[{"commit":"10f341af44e56f6dad87174da19746ca5bbccd9f","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/10f341af44e56f6dad87174da19746ca5bbccd9f"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-11 14:42:37.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-12 13:14:06.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Mistral service.\n\nChange-Id: I574c8233e15240074e67520b30df9cca79a9c20b\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/82cd962e78cab57395f61abacee6d6502067fc99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/82cd962e78cab57395f61abacee6d6502067fc99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
