)]}'
{"id":"openstack%2Fpuppet-mistral~746182","triplet_id":"openstack%2Fpuppet-mistral~stable%2Ftrain~I574c8233e15240074e67520b30df9cca79a9c20b","project":"openstack/puppet-mistral","branch":"stable/train","hashtags":[],"change_id":"I574c8233e15240074e67520b30df9cca79a9c20b","subject":"Expose vhost_custom_fragment","status":"MERGED","created":"2020-08-13 16:01:16.000000000","updated":"2020-08-14 10:21:07.000000000","submitted":"2020-08-14 10:21:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746182-1597400467014-c5357c5d","meta_rev_id":"d96563348356952e1408432e55ebfbaa71b54163","_number":746182,"virtual_id_number":746182,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 10:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-13 20:10:16.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-14 09:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-14 09:06:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 20:10:16.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-14 09:06:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-14 10:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86dcdb7263da65959f3ea5fc04c10dd7612dd70e","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-13 16:01:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db58872989fcd045c37e85d3f07ea01dd00e752c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 17:28:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73e917ad65494183b974b52f2170add1 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/851a9621e2e44ea7ba26a0023c798228 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b778ffb7c1f9466a80e0484d8db4f36f : SUCCESS in 6m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f8187150d4cf42a2ae8bd80f27087381 : SUCCESS in 13m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d37e0fc4cfba4a8b9936596042568a0c : SUCCESS in 12m 33s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/f607ae3d10f9444f9a3694873bd86757 : SUCCESS in 13m 58s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48fea0576711427291988486eecd3267 : SUCCESS in 12m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50431bbb87af47cea17b9f5bd823a3c4 : SUCCESS in 11m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eeb077eae2d3467fb191751ccffe2020 : SUCCESS in 42m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9d80ae3259514d10ba62ad44cee67aeb : SUCCESS in 55m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4774fbdb0e5c451382cdd722755aab9c : FAILURE in 8m 42s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2aab9a50ecd0416fb17ea72467b089ed : SUCCESS in 24m 15s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b4c323bcf86424980eb8708bb7e0a5b : FAILURE in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29968040a78fed68f3e054a78c2c6a6749272381","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-13 20:10:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb870321b0c6737bd6936752cd1da0f89f098884","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-14 09:06:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e3134743fcf3544c9c458acf914aa2d2905ec91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 09:06:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37ed0e994eb6541549efd2538468c4da746582fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:21:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d56f199354734d9490557ccbf9b3ac50 : SUCCESS in 6m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44bfdbbff350426393a4776513206989 : SUCCESS in 6m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d7afeb2717a4e5ea538527c9caa84e3 : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2d6c04585e38458e8409588727f4d870 : SUCCESS in 19m 05s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35f33f7fc2ee439d82e61dd2ec4b0637 : SUCCESS in 14m 21s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a24627320d944fc4b128c27afd0d0171 : SUCCESS in 1h 05m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a6a8cc5627854903b0f7d359b8a1ff52 : SUCCESS in 22m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d96563348356952e1408432e55ebfbaa71b54163","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:21:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"426952d1259c3e24fbb3618278cc7f4e4f9e8ebc","revisions":{"426952d1259c3e24fbb3618278cc7f4e4f9e8ebc":{"kind":"REWORK","_number":1,"created":"2020-08-13 16:01:16.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/82/746182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/82/746182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/82/746182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/82/746182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/82/746182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/82/746182/1"}}},"commit":{"parents":[{"commit":"c733b8a16aca539a88e9c7a28631265490af4691","subject":"Switch to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/c733b8a16aca539a88e9c7a28631265490af4691"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-11 14:42:37.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-13 16:01:02.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Mistral service.\n\nChange-Id: I574c8233e15240074e67520b30df9cca79a9c20b\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n(cherry picked from commit 82cd962e78cab57395f61abacee6d6502067fc99)\n(cherry picked from commit 3c95d185ca3a82367a0572674eaccb94e5ad6de8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/426952d1259c3e24fbb3618278cc7f4e4f9e8ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/426952d1259c3e24fbb3618278cc7f4e4f9e8ebc"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
