)]}'
{"id":"openstack%2Fpuppet-mistral~747090","triplet_id":"openstack%2Fpuppet-mistral~master~Ibd8a469428cf99cdafb3668e00c3b05b4762e7eb","project":"openstack/puppet-mistral","branch":"master","topic":"bug/1892284","hashtags":[],"change_id":"Ibd8a469428cf99cdafb3668e00c3b05b4762e7eb","subject":"Add service_token_roles for keystone authtoken config","status":"MERGED","created":"2020-08-20 02:42:48.000000000","updated":"2020-08-20 23:52:29.000000000","submitted":"2020-08-20 23:51:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747090-1597967465193-7da3442f","meta_rev_id":"b7adcba8270dd17991b697c5c326078ca13b7651","_number":747090,"virtual_id_number":747090,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-08-20 20:51:57.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 23:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-20 12:11:05.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-20 14:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-20 14:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 12:11:05.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-20 14:33:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-08-20 20:51:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-20 23:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"546fb0beb985343020abdcfac00827617da954c7","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-20 02:42:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9f07221ed2672f9012647e0094794c90c049b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 04:25:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc998ec17c2f4e268950f3e1cd3363a9 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67995345b30941fba57a14570dad2dfa : SUCCESS in 5m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9a409219fa94d68a016fa1abc2c97e7 : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/00115b271cd5496d81b968700dd23738 : SUCCESS in 11m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1fd099a1eea04188a2e7ae5098423d49 : SUCCESS in 12m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02af9570d8304867ad749927a5503375 : SUCCESS in 11m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e7485f986314ea78d5e6557f1f42c64 : FAILURE in 7m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73e10325fd054bb4bb506c33061937fb : SUCCESS in 40m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/640019c9f03b45028e16e79247903e4b : SUCCESS in 1h 28m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5f7c3e4d431441f48ef0aab75d7f8a5f : SUCCESS in 20m 20s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c648cf5eb55440bda6a480f29abdf94c : FAILURE in 8m 04s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2840f2fa32e34ed89da061900bfdc508 : FAILURE in 8m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a9b75b51118488298d2b1624ba8163c : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"deb2d73cabaff1e7ce53192c5afb17d19fe6c684","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-20 12:11:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"045d8cfe7d74f82fb1c007e7f5bfdafa2ed9fc8f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-20 14:33:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d48c6bc5189332561e5906db63ec020bcdeaecb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 14:35:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18503f9b7fc03bf03684078171a4c834dde6b015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 16:14:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5f60988d1c54033a8ffb1c9d5a54897 : RETRY_LIMIT in 1m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/761df46414d140dfb000c9a3c434dd44 : RETRY_LIMIT in 1m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f1dc949237c49e1a348540ec026d383 : RETRY_LIMIT in 1m 32s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f57ae8aba3f5480daf03eb978d49c5d2 : SUCCESS in 13m 39s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/dec00df5cdb44891a4338219ccdd3535 : SUCCESS in 12m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d88c48ed76b5489f9874866e4f477e3a : RETRY_LIMIT in 1m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/996b955295844b2397980d95f67f6b39 : SUCCESS in 1h 22m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1d88b604fbd142b987e7be8217963c69 : SUCCESS in 20m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31c4d7e5272a45b3ac4921ec2aad4610 : RETRY_LIMIT in 1m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d14fc1735d76c0b94eb2d44562283d8ed62d1d0f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-20 20:51:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c7f0f6a3f019b6ce7a02cdc06e3d8a673b099c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 22:25:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9caa2b8892104989814a97e8d7ee08ca : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9294a9c854bf41b0a15d5ec5a79f935b : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ee1c0d04c9341928b34931d2f8b1d63 : SUCCESS in 5m 35s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/85e8b86138e34f5f8655d843b5c466ee : SUCCESS in 14m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9fe564935b8245828f60bbfb7f8a74d5 : SUCCESS in 15m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38b5e0f264c04b89bd7760a944c07bfe : SUCCESS in 12m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0b36f35942b4a4b99fa04c1e0b9fbbe : FAILURE in 10m 38s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f184f2f3091d4d36ac6feddb63789d6f : SUCCESS in 46m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ab59a7eefc614850966fe22657389f26 : SUCCESS in 1h 23m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d58ad58346264b11bdb79dbfd0a78089 : SUCCESS in 18m 51s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/aeffb69f214644c0b414e6e009d13907 : FAILURE in 10m 44s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b897c44e639f46f79606af5735246127 : FAILURE in 11m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/865f5ab8eac14b87bf39cf493bb5a995 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c44048b9feb2d455b6912d15b50f80f3a3181756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 22:25:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca98d29b82e5ddb7a7d510636b8a8e5b33a5a8ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:51:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d2ac11c0a37465ab33e571fc7368a12 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b57a5fc390c545a487b6cff8d3c6ea40 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7730ed515d2a4df89eb15ef7d9b8b46a : SUCCESS in 4m 59s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bd3b3ba664064b51b4e3f4b3f5b247c0 : SUCCESS in 12m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/bc24b05788094f26804413917bf2317b : SUCCESS in 15m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/420dad449092444aa396ea6a10c69a83 : SUCCESS in 10m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e4b1b226dfb4e94bea5f3ec1b765ddc : SUCCESS in 1h 22m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9b111e9bd2b0467e882fdc5b19dd323b : SUCCESS in 20m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/105323dc8217484e85de9d6cc697380b : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c8816cfad7c579f9d3abc0d2ac2d613c2d6e9a95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:51:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f654e7f25cf2dc4a0c3965a182195f8b104676ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:52:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6c00101453a40ee90b3d81b07d0730e : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e361bf81816c130bedf3b0c96385108fee1729a3","revisions":{"e361bf81816c130bedf3b0c96385108fee1729a3":{"kind":"REWORK","_number":1,"created":"2020-08-20 02:42:48.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/90/747090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/90/747090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/747090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/747090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/90/747090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/90/747090/1"}}},"commit":{"parents":[{"commit":"82cd962e78cab57395f61abacee6d6502067fc99","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/82cd962e78cab57395f61abacee6d6502067fc99"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-20 02:41:27.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-20 02:41:27.000000000","tz":480},"subject":"Add service_token_roles for keystone authtoken config","message":"Add service_token_roles for keystone authtoken config\n\nAdd the ability to configure service_token_roles.\n\nChange-Id: Ibd8a469428cf99cdafb3668e00c3b05b4762e7eb\nCloses-Bug: #1892284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/e361bf81816c130bedf3b0c96385108fee1729a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/e361bf81816c130bedf3b0c96385108fee1729a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
