)]}'
{"id":"openstack%2Fpuppet-mistral~748938","triplet_id":"openstack%2Fpuppet-mistral~stable%2Fussuri~Ie4fea38fb96d1c04a8a589f2f216a167fd15b827","project":"openstack/puppet-mistral","branch":"stable/ussuri","topic":"db_conn_default-stable/ussuri","hashtags":[],"change_id":"Ie4fea38fb96d1c04a8a589f2f216a167fd15b827","subject":"Fix the default value of mistral::database_connection","status":"MERGED","created":"2020-08-31 02:02:42.000000000","updated":"2020-08-31 15:03:15.000000000","submitted":"2020-08-31 15:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748938-1598886195464-f4e8b138","meta_rev_id":"7f0673b5a15e44d39f04086ab30cfb6925f40b7d","_number":748938,"virtual_id_number":748938,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-31 15:03:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-31 13:28:16.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-31 07:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-31 13:28:16.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 07:48:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-31 13:28:16.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-31 15:03:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e236e12456f12c6fb746ece85e76de8ef66ce4d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 02:02:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1e47c542b8572d2c7b96c5279504a08cbbf88210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 03:14:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f511a69aaaa48f18d9c8c821c51e83c : SUCCESS in 4m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68fb7caec8ca4b02829e97b5b17bb171 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c0d7bf9c0c04ec782371a45550e105c : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f4cfebc3b194490598c33ba5032c5aaa : SUCCESS in 11m 21s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2567a8dedaee4655975bbd10b6e21895 : SUCCESS in 11m 03s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adced4412c0a4d9e86a63dfd2b0b7dc1 : SUCCESS in 7m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9263fb504804f85aab3c5d981f7176f : SUCCESS in 12m 48s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16e074be887a45238daf39d8230f9231 : SUCCESS in 43m 29s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f40ed100a376455ebd8dbdd3c1faf50b : SUCCESS in 1h 10m 12s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2ce88ebf6cfb404d956a6eae555fc4a5 : SUCCESS in 17m 08s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/45bb7160a48c4186866f8a7d4a1f91f3 : FAILURE in 8m 34s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7e9d61d949c4860b1fba3f13ba4014d : FAILURE in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"203935a58244d86109c567f357517099d26907c2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 07:48:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6243bf22753509fc83d7fe1bc046c14de962a518","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-31 13:28:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"124ae55dec657dffaf6e966ed0eec49eb6063398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:28:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37fb99abaf04e856df11f9dff31f3d51bda7e862","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:03:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8efd69b3a89a616ea98ca0d33ed1ca4883f4f305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:03:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/874c9bc69e164ae78ab15183df4b0fb2 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd86a440f6234f41b6cc3ad2f245c668 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20d837ccd6e64d81b346451d17501b09 : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1d72eb51a14043519aba050cb0e1ad2b : SUCCESS in 15m 21s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/90e681078d4a414989d15a981bd353a1 : SUCCESS in 13m 44s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf977578bc464c9784ae0f60e9b57f3b : SUCCESS in 10m 53s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7bf30ae04ceb4123b0c25eb63aa014a9 : SUCCESS in 1h 06m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e019f2efbf3b4515bb7a69bfeafb809f : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b04597ed60d6261988601dfbfb3239dbcd8cf5a9","revisions":{"b04597ed60d6261988601dfbfb3239dbcd8cf5a9":{"kind":"REWORK","_number":1,"created":"2020-08-31 02:02:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/748938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/38/748938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/38/748938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/38/748938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/38/748938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/38/748938/1"}}},"commit":{"parents":[{"commit":"7f53aa1e64849003f7bfb9c6b9d826141785f0b1","subject":"Add CentOS8 to nodesets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/7f53aa1e64849003f7bfb9c6b9d826141785f0b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-30 14:49:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 02:02:42.000000000","tz":0},"subject":"Fix the default value of mistral::database_connection","message":"Fix the default value of mistral::database_connection\n\nThe default of mistral::database_connection should be undef instead of\n$os_service_default, otherwise mistral::db::database_connection,\ndoesn\u0027t work at all.\nThis patch fixes the default value so that it has the corret one\ndescribed in parameter description.\n\nChange-Id: Ie4fea38fb96d1c04a8a589f2f216a167fd15b827\n(cherry picked from commit 201131e1cbd00c7576b74f0a512ddc8949cf21d6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b04597ed60d6261988601dfbfb3239dbcd8cf5a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b04597ed60d6261988601dfbfb3239dbcd8cf5a9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
