)]}'
{"id":"openstack%2Fpuppet-mistral~749120","triplet_id":"openstack%2Fpuppet-mistral~master~I85f54c041eceb67611c476472391ff1e44dfc8eb","project":"openstack/puppet-mistral","branch":"master","hashtags":[],"change_id":"I85f54c041eceb67611c476472391ff1e44dfc8eb","subject":"Add Puppet Litmus","status":"MERGED","created":"2020-08-31 19:37:24.000000000","updated":"2020-09-10 02:31:58.000000000","submitted":"2020-09-10 02:31:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749120-1599705118623-c09e01d6","meta_rev_id":"63502742691b8f311723d8dbcc97320567db4afb","_number":749120,"virtual_id_number":749120,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-09-09 22:56:18.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 02:31:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 18:28:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-08 18:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 18:28:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-09 21:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 18:28:03.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-09 21:27:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-10 02:31:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"018a9a7719400aca199de7c25bf11d78cba8290b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 19:37:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb4f3708cae54ad9f279b529031d6f8ffa2e725c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 21:08:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/516d840e7f0d45178601ab63f2ec694d : SUCCESS in 3m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/beb39cf7d937448d8e5c5db938df0f71 : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5256b594e3ac417da1e6436db048e4fe : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90953b8af97f4d27a6bc316d5743357d : SUCCESS in 7m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e15972212a6d4686874b09059a96115d : SUCCESS in 8m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22962e1bc3614cc88052c3a1c12f2756 : SUCCESS in 43m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/13cf7f8c7e1c475cb5abee8ca8ff0530 : SUCCESS in 1h 25m 01s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a14d3f31b78b484e99a2c9654d2ddfe4 : SUCCESS in 21m 10s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99530909791e48d489f90b2d839feccd : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"38ce69015c03cecd5cc1fc5083c72a832804682a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:20:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d08975d7493229f2a1fc9bbf8fda7645f074a4df","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 18:28:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aed5a54ee98c75b64cf010d4d6d984f0a92baf32","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-09 21:27:21.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c37d8db90516853cfaf365e60445f14ba8d08ad0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-09 22:56:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"660c0a633f7d17487a62f15727cae588c5ff6399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:07:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f41728410d524c4287846aada5c81d6a : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72f2a0f5f5d64e358010d8fde1f33872 : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/37a1270897724339a2f49fedc888d0ab : SUCCESS in 10m 26s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a82120d7a6c4409860326c8823e71e3 : SUCCESS in 13m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f91ff331ebc422c9112abb3ebe21da1 : SUCCESS in 9m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/406328baf36e40e6a79580282378c8b1 : SUCCESS in 1h 40m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b1874258ffa5498fb1437c77ce547885 : SUCCESS in 1h 20m 35s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/61ce8affb223402faac8d8a9c1f36b07 : SUCCESS in 20m 32s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/473bca183582412095030e0bb0799b0c : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"790624cf9025e562a9e92939e539fa513c9aa84c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:07:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f846374ceff1ca38b7fc92167d8fb05a8b4e4e01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 02:31:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63502742691b8f311723d8dbcc97320567db4afb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 02:31:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cf126fc9485449e93659a6a6275acfd : SUCCESS in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cd61109d3c946ffa8cb6e07a70c3bbe : SUCCESS in 3m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4e9ddb07130f4f0b855c994f7115e1b3 : SUCCESS in 8m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7428252db334e25b8d0463f869ad344 : SUCCESS in 8m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fef8af8413f147de8bfc7e7c8732f5da : SUCCESS in 1h 20m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d66539f2d5d548edb723039f94ecab03 : SUCCESS in 19m 32s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb8d40b172fd4a14866ee9bf8218e183 : SUCCESS in 15m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b437053e405d9c3d57afb2627b1dd27653441b52","revisions":{"b437053e405d9c3d57afb2627b1dd27653441b52":{"kind":"REWORK","_number":1,"created":"2020-08-31 19:37:24.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/20/749120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/20/749120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/20/749120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/20/749120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/20/749120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/20/749120/1"}}},"commit":{"parents":[{"commit":"af025be9db7bde3e6a374028d6d472d8706946fd","subject":"Merge \"Use openstack_spec_helper from zuul checkout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/af025be9db7bde3e6a374028d6d472d8706946fd"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:37:17.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:37:17.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: I85f54c041eceb67611c476472391ff1e44dfc8eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b437053e405d9c3d57afb2627b1dd27653441b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b437053e405d9c3d57afb2627b1dd27653441b52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
