)]}'
{"id":"openstack%2Fpuppet-mistral~753918","triplet_id":"openstack%2Fpuppet-mistral~master~Iba84720c6cb22e6f6ffd6a3df1778d0740be03fd","project":"openstack/puppet-mistral","branch":"master","topic":"prep-release-victoria-rc1","hashtags":[],"change_id":"Iba84720c6cb22e6f6ffd6a3df1778d0740be03fd","subject":"Prepare Victoria RC1","status":"MERGED","created":"2020-09-24 02:49:54.000000000","updated":"2020-09-25 12:48:06.000000000","submitted":"2020-09-25 12:48:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753918-1601038086564-b2ad87a5","meta_rev_id":"5653bc65575684d13bf9eacc515337025c47f067","_number":753918,"virtual_id_number":753918,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 12:48:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-25 12:37:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-24 11:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-25 12:37:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 11:40:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-09-25 12:37:24.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-25 12:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0868a4624e622a92f7105f9ecf2d93671f1f113c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-09-24 02:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81d745af3ec6d37378b6c468de3ac2ee169f2f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:58:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/9bf651e35f6e40e5a65168e80b0b85e0 : SUCCESS in 2m 05s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a36765bc24f24d60b298ff20c7ba2238 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"db10a6096f5b2fe952406356053ec32cc1518b87","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-24 11:40:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"13b0727488fccb0bbb607c015b9b6a9edc7c83ea","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-25 12:37:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be061d2620ce95ce91c2ac9743e04e4bf0a242a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:37:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7aa4e1fbcc38fd3179701255ab7477763552436","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:48:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"de63eae96f29ef69524ecb280629f63e6cef9f77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:48:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/6be9e6ebe49e4ee298ea6db6df71a3ca : SUCCESS in 2m 24s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58d3a0adf90d4a91b3cac35507cb8f67 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c08d21dc7d65234e3de0e1768eee04d94d7817e9","revisions":{"c08d21dc7d65234e3de0e1768eee04d94d7817e9":{"kind":"REWORK","_number":1,"created":"2020-09-24 02:49:54.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/18/753918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/18/753918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/18/753918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/18/753918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/18/753918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/18/753918/1"}}},"commit":{"parents":[{"commit":"6a4b4a6a2561b2e25ad91f2fe699a3d0de89eb8c","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/6a4b4a6a2561b2e25ad91f2fe699a3d0de89eb8c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-09-24 02:48:54.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-09-24 02:48:54.000000000","tz":480},"subject":"Prepare Victoria RC1","message":"Prepare Victoria RC1\n\nUpdate the version metadata for Victoria milestone RC1\n\nChange-Id: Iba84720c6cb22e6f6ffd6a3df1778d0740be03fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/c08d21dc7d65234e3de0e1768eee04d94d7817e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/c08d21dc7d65234e3de0e1768eee04d94d7817e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
