)]}'
{"id":"openstack%2Fpuppet-mistral~791374","triplet_id":"openstack%2Fpuppet-mistral~stable%2Ftrain~I0ad151db5805c08cd1ad8096a755d916a61a6a67","project":"openstack/puppet-mistral","branch":"stable/train","topic":"t-release","hashtags":[],"change_id":"I0ad151db5805c08cd1ad8096a755d916a61a6a67","subject":"Prepare the final stable/train release","status":"MERGED","created":"2021-05-14 06:56:12.000000000","updated":"2021-05-23 13:10:34.000000000","submitted":"2021-05-23 13:10:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791374-t-release","meta_rev_id":"8f156200f8ce7ef850c53c5b64187b64f5de17c6","_number":791374,"virtual_id_number":791374,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-23 13:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-05-17 15:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-05-17 18:22:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-05-17 18:22:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 07:03:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 15:47:04.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-17 18:22:11.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3a07ec4e76d09e30467f613c1bd03f2dd8b18617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-14 06:56:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"959a1d89d1b48b03d7253b0f7846a182b712d883","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-14 06:56:54.000000000","message":"Topic set to t-release","accounts_in_message":[],"_revision_number":1},{"id":"6a25fbf36d3e965dc88ccd1c314cc172db98d5f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 07:03:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ccb780516d544e41963fa45161920f63 : SUCCESS in 4m 29s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cb9835142004cfd8db1d80d2565e67e : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"893182bc54983d7d7b4817c0fa14a550bc67bd99","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-17 15:47:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f06b1660ce41c5d2a049d92d2cc2d76d6ba85107","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-17 18:22:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b34f89cb13cca25c9e3d3b60809b370189136aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-23 12:59:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0f9a3f3a6db37b16a1e985fe6b726ca3367a121a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 13:05:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2e9232c7d3864d8fb33b89c90ecee250 : SUCCESS in 3m 08s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2934e97d7ca4e6cb5ddd34ffc5b8842 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"731703a5e09b7d9e264cea4c0bb1ef92c516837f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 13:05:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2e83d1f17bf8ef396ab9154b027d4e7d1093dee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 13:10:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/72a59e34d62b44a5ab2f8cf4d2dc7795 : SUCCESS in 2m 25s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1497aff5ff5c44fc8943e248e3014bbb : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8f156200f8ce7ef850c53c5b64187b64f5de17c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 13:10:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5dcd2371f39ae151bc9b0f10d0ce9e1a52746860","revisions":{"5dcd2371f39ae151bc9b0f10d0ce9e1a52746860":{"kind":"REWORK","_number":1,"created":"2021-05-14 06:56:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/791374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/74/791374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/74/791374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/74/791374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/74/791374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/74/791374/1"}}},"commit":{"parents":[{"commit":"426952d1259c3e24fbb3618278cc7f4e4f9e8ebc","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/426952d1259c3e24fbb3618278cc7f4e4f9e8ebc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-14 06:55:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-14 06:55:58.000000000","tz":540},"subject":"Prepare the final stable/train release","message":"Prepare the final stable/train release\n\nThe stable/train is entering in extended manitenance. Because we have\nseveal changes merged since the last release, a new release should be\ncreated before that.\n\nDepends-on: https://review.opendev.org/791063\nDepends-on: https://review.opendev.org/791064\nDepends-on: https://review.opendev.org/791204\nChange-Id: I0ad151db5805c08cd1ad8096a755d916a61a6a67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5dcd2371f39ae151bc9b0f10d0ce9e1a52746860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5dcd2371f39ae151bc9b0f10d0ce9e1a52746860"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
