)]}'
{"id":"openstack%2Fpuppet-mistral~797449","triplet_id":"openstack%2Fpuppet-mistral~stable%2Fussuri~Ia5dae538a30ad32e73cd5121fdcc5bc14a6aece2","project":"openstack/puppet-mistral","branch":"stable/ussuri","topic":"797667","hashtags":[],"change_id":"Ia5dae538a30ad32e73cd5121fdcc5bc14a6aece2","subject":"Test 797447","status":"ABANDONED","created":"2021-06-22 13:33:27.000000000","updated":"2021-07-07 03:50:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0a7e177f702290450c0098a6d2e95da859514c76","_number":797449,"virtual_id_number":797449,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-22 14:03:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-22 14:03:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed202b29e1efc44547db05c1c357d578fe23c1da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-22 13:33:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"446344b8bda2c0abcb84c9ad89f685e9b8a35892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 14:03:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6da5b1e472c422b8babd85c71a9f93d : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e45bc1f5fa24e92a768af7ddf7bbf8b : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22e89e17e48b4bbeb6167810268ca94c : SUCCESS in 4m 42s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/868d8a64722a402d9638d8d017419838 : FAILURE in 22m 25s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/61ca5d58a9d743ee92c178267ab27642 : FAILURE in 26m 31s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2073cec49f1c44ae8b10d6bc1a450ffd : SUCCESS in 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7f54542cee6a08aa839c080876037480535fba6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-27 05:08:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"98b521277c47a1169d715dbbaaf7439821cef416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-27 05:45:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6d005a42fd34c9f9be6150166e38e28 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9a01c657a7a468287c5a957c1ce6dde : SUCCESS in 7m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16b0466ac7cc430fae818e51221bef7a : SUCCESS in 5m 27s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/669e284cd7cf4d82a7008ae1a9a06352 : FAILURE in 33m 28s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/554122abbb7a406b8727a2efdbfc2835 : FAILURE in 26m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2aa902ae383042a1b8d41d79dfa3fb7b : SUCCESS in 19m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a7e177f702290450c0098a6d2e95da859514c76","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-07 03:50:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c1de58dfbf995bc0b3472a39df5505daac190b5","revisions":{"2c1de58dfbf995bc0b3472a39df5505daac190b5":{"kind":"REWORK","_number":1,"created":"2021-06-22 13:33:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/797449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/49/797449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/49/797449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/49/797449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/49/797449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/49/797449/1"}}},"commit":{"parents":[{"commit":"4711214d0932cfcbfc5501d3b36e57fca1e06cab","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/4711214d0932cfcbfc5501d3b36e57fca1e06cab"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-22 13:32:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-22 13:32:56.000000000","tz":540},"subject":"Test 797447","message":"Test 797447\n\nDepends-on: https://review.opendev.org/797448\nChange-Id: Ia5dae538a30ad32e73cd5121fdcc5bc14a6aece2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/2c1de58dfbf995bc0b3472a39df5505daac190b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/2c1de58dfbf995bc0b3472a39df5505daac190b5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
