)]}'
{"id":"openstack%2Fpuppet-mistral~799575","triplet_id":"openstack%2Fpuppet-mistral~master~I5c2653475bbbf6dc0a9ce795958fd884bcdf13e8","project":"openstack/puppet-mistral","branch":"master","topic":"coordination","hashtags":[],"change_id":"I5c2653475bbbf6dc0a9ce795958fd884bcdf13e8","subject":"Use oslo::coordination to manage coordination parameters","status":"MERGED","created":"2021-07-06 08:03:48.000000000","updated":"2021-07-07 21:27:14.000000000","submitted":"2021-07-07 21:25:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799575-coordination","meta_rev_id":"3efa0927883b28e2256b44f7aeca5e3ecdd0897f","_number":799575,"virtual_id_number":799575,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-07 21:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-07-07 20:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-07-07 20:11:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-07-07 20:11:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 09:41:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 20:06:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-07 20:11:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"bea41960d03005625206858045eb84ae6c3c9bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-06 08:03:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ad23bdbea07193c5e958255ac31f5d1f1ffd613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 09:41:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/011399c1d3fd4788935b94b915a89d1f : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5ed04d4f59f444abb8ccec1c128b135 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76ceb4fbe33f487097ed493ed46ddd47 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/be10a7635cec4b1ba3f0dfa4dcdfc614 : FAILURE in 10m 38s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c34ef1db1084391a07c0f6031e39bbc : FAILURE in 11m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d93d6a25e32496aabda1781ab047bc2 : FAILURE in 7m 57s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b31bfcb1e07f4ea194633cd16e602aa7 : FAILURE in 8m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b09d266a33fa4281a9188989a0a75f63 : FAILURE in 8m 54s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b58c3442ade34289b16b7638eda33fad : FAILURE in 29m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d26b389440d8406a8e930349197d1984 : SUCCESS in 1h 25m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a7d74680b6649eaa98fd09455983893 : SUCCESS in 1h 27m 45s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7d2a15524434a07936a610dd00156c8 : SUCCESS in 27m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4fc73753ab94b1b8c7ae6b6471bbf1b : FAILURE in 16m 47s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2b5a61a98c34ffda9af4619628ec871 : SUCCESS in 27m 20s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04a24d268ff04e00addfd342583e74ca : FAILURE in 16m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24060c213ff443bbab9fbf2bac61032c : FAILURE in 16m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4704a093f3b4c88953560601139acea : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"eb20c015676a5107dee24911364878b3c5700240","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-06 09:49:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8a5fa72f29a238af8692a9a17433b55068cf263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 11:23:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45207d46220944f48ff79f38c9747587 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/748a77841be148ed9eb15a18e5a03dac : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae83a33b5ec14dcf9a8ed6798afd54f5 : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/905c35eae7a64c1fbade36019e2dbe89 : FAILURE in 11m 44s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/869c48793c564917b03b656cb7405417 : FAILURE in 12m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d182b0ad06f24084b4bc26af8ac012c2 : FAILURE in 9m 04s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ad9969060bc47f49d1207e18c7f6d3b : FAILURE in 7m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d4b056989cf4fe8bc5d43087401f6b4 : FAILURE in 8m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33c90e9b4051405fb8a00faa9e74f197 : FAILURE in 30m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cca4aebce50d49039be68f4adddefe4a : SUCCESS in 1h 29m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a8cbfc97d5a40fa91850fe21850a5e5 : SUCCESS in 1h 28m 14s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e560af14c5c84ec59f02322f8402cb5f : SUCCESS in 24m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0416e5a96834ef7a3442303ad389610 : FAILURE in 16m 23s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/45b424561afe46bf82b4c5828acb3a3c : SUCCESS in 24m 14s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5efa5e2f286143639aeb68ae645a88a8 : FAILURE in 17m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7effdcd27b104efbb6bc8641587757a3 : FAILURE in 16m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbbe35a576d24ca5ac01d0ff41c036a1 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a39e54e2d78963d2a0d7920268070e7a9356841c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-06 11:38:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3625ce36f2598c67e0096658e8bddd543fbc7619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 12:52:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a3d95d7918a4c60a906937d03cf5b05 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bab8f318d8f64ea8aef7622da7b61027 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fca3b6a291b34d589d7539468babf757 : SUCCESS in 4m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0857a29bd73149d7bb7a787a03eb834a : SUCCESS in 11m 39s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e2c981400ab4c6e92acc615a4bcf40f : SUCCESS in 12m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c6494f9db4d4bf1b096b924f6ea6962 : SUCCESS in 8m 17s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c012c3cc7d24bcd8ff3e8501682cab7 : SUCCESS in 8m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10232fdf406141a8aff767c9afa24f62 : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7d3d08679774278947e211343b10afd : FAILURE in 28m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fba0b485c2894543a3f9a7db0ca5dfdb : SUCCESS in 1h 12m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfcc9efe3a71463dbec0413d9c6feedf : SUCCESS in 1h 10m 25s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e08f2b9711e440db491b04be172874e : SUCCESS in 25m 45s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c7ad9f0a95b49b7921be5332d4fe340 : FAILURE in 18m 23s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/880afd3ca1e942d9aa0cd76748041d58 : SUCCESS in 25m 14s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/863f745477fb4ac8b0011ede1a9f1187 : FAILURE in 17m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b92624daea14508afcaa40a980e0c32 : FAILURE in 17m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42229c73956c4042a52f17eb1cd695dc : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"5a65a12b9a89da2564cfdbb7014ab86cf93ae5c0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-07 20:06:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8db47129edad4a96d2fc2f5d5068ed9b8bce59ef","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-07-07 20:11:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e21984ac6fb2920ad4d32ab94f62aece530f148e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:12:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b1d1e8260a6c4e7c03acbbfd33d20330c81bf129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 21:25:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e26a17af2be47b98b1b3da8cf68b825 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22439334f22e4f3bb6039d6611957881 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bc85cc412924c1d885ee4fad7efbcc8 : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0d9f78736934ef0b78529d414e6973c : SUCCESS in 12m 46s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/62d0168a39774c95835688349c2ee65e : SUCCESS in 13m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef86072c8fa74de2801541537fae6eaf : SUCCESS in 9m 35s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b2e624f5f4240c895ab86062eee3937 : SUCCESS in 10m 52s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fe4a2874714487884ee45162643e178 : SUCCESS in 1h 12m 30s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/501c245f9a374c6f9710d69033961b19 : SUCCESS in 26m 57s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e3ec2d1cae8497e91f98864d27d1907 : FAILURE in 16m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e12866993d0449218ab553335d22726d : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":3},{"id":"309efade6487819b582c68cd217984a2e2c7da30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 21:25:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3efa0927883b28e2256b44f7aeca5e3ecdd0897f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 21:27:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae96c06a4af445feaa50191ec13162cd : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"298c6ec174780d97966d1724107910ab88d26def","revisions":{"92b545e73698ca2c69db98fee4bc7fea64f5a5a4":{"kind":"REWORK","_number":1,"created":"2021-07-06 08:03:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/799575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/75/799575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/1"}}},"commit":{"parents":[{"commit":"b89a7dc5bc589f1079daefbdbd94193f3001d376","subject":"Merge \"Do not test detail of oslo::cors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b89a7dc5bc589f1079daefbdbd94193f3001d376"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 08:03:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 08:03:22.000000000","tz":540},"subject":"Use oslo::coordination to manage coordination parameters","message":"Use oslo::coordination to manage coordination parameters\n\nThis change replaces current implementation to manage coordination\nparameters by the base class in puppet-oslo. With this change\nthe required backend package is installed according to the backend\ndriver used.\n\nDepends-on: https://review.opendev.org/791628\nChange-Id: I5c2653475bbbf6dc0a9ce795958fd884bcdf13e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/92b545e73698ca2c69db98fee4bc7fea64f5a5a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/92b545e73698ca2c69db98fee4bc7fea64f5a5a4"}]},"branch":"refs/heads/master"},"14ff00189551b70716225cfb0825757c2d067d46":{"kind":"REWORK","_number":2,"created":"2021-07-06 09:49:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/799575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/75/799575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/2"}}},"commit":{"parents":[{"commit":"b89a7dc5bc589f1079daefbdbd94193f3001d376","subject":"Merge \"Do not test detail of oslo::cors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b89a7dc5bc589f1079daefbdbd94193f3001d376"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 08:03:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 09:48:59.000000000","tz":540},"subject":"Use oslo::coordination to manage coordination parameters","message":"Use oslo::coordination to manage coordination parameters\n\nThis change replaces current implementation to manage coordination\nparameters by the base class in puppet-oslo. With this change\nthe required backend package is installed according to the backend\ndriver used.\n\nDepends-on: https://review.opendev.org/791628\nChange-Id: I5c2653475bbbf6dc0a9ce795958fd884bcdf13e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/14ff00189551b70716225cfb0825757c2d067d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/14ff00189551b70716225cfb0825757c2d067d46"}]},"branch":"refs/heads/master"},"298c6ec174780d97966d1724107910ab88d26def":{"kind":"REWORK","_number":3,"created":"2021-07-06 11:38:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/799575/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/75/799575/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/75/799575/3"}}},"commit":{"parents":[{"commit":"b89a7dc5bc589f1079daefbdbd94193f3001d376","subject":"Merge \"Do not test detail of oslo::cors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b89a7dc5bc589f1079daefbdbd94193f3001d376"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 08:03:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 11:38:08.000000000","tz":540},"subject":"Use oslo::coordination to manage coordination parameters","message":"Use oslo::coordination to manage coordination parameters\n\nThis change replaces current implementation to manage coordination\nparameters by the base class in puppet-oslo. With this change\nthe required backend package is installed according to the backend\ndriver used.\n\nDepends-on: https://review.opendev.org/791628\nChange-Id: I5c2653475bbbf6dc0a9ce795958fd884bcdf13e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/298c6ec174780d97966d1724107910ab88d26def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/298c6ec174780d97966d1724107910ab88d26def"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
