)]}'
{"id":"openstack%2Fpuppet-mistral~808887","triplet_id":"openstack%2Fpuppet-mistral~master~I531c77be952daf3f0e86185b9588307da2d1e492","project":"openstack/puppet-mistral","branch":"master","topic":"bug/1943212","hashtags":[],"change_id":"I531c77be952daf3f0e86185b9588307da2d1e492","subject":"Add watch_log_file option","status":"MERGED","created":"2021-09-14 08:02:55.000000000","updated":"2021-09-15 09:50:22.000000000","submitted":"2021-09-15 09:48:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808887-bug/1943212","meta_rev_id":"ee64017d9c8a2a55585704fd354428a95e5b5458","_number":808887,"virtual_id_number":808887,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 09:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-09-15 01:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-09-14 15:11:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-09-15 01:12:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-14 09:42:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 15:11:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"30271bc11696ba8569481a2c310dbd7d5c58fc4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-14 08:02:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b22235363db879e6556d5e1c6a5fce5799bccd77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 09:42:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df7e600baba949f0920f6f7cc2d3c5ee : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/078c5ac84b364ecaa9016369bcee83f8 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/786d9f5b0bbc4a4d91d9612770737323 : SUCCESS in 4m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a95af385414e41708b0318e2833792c2 : SUCCESS in 13m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccabab384a674a83b0fde96e7cbdfd64 : SUCCESS in 12m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11c7576e60e1401684233c66a69d23f1 : SUCCESS in 10m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/360d30b0dc394408aa2c0dbce1607844 : SUCCESS in 10m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f70db28772384f74873e6f5a2adf7af9 : SUCCESS in 12m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da2f12143cd64f0e93b40bef5668a991 : FAILURE in 35m 35s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b09b6149f7214229987579845b3c3589 : SUCCESS in 1h 32m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c390025ece349bc9a6e982ceb66fcaf : SUCCESS in 17m 12s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2464bd90738945e3a241a3fcc4914fca : SUCCESS in 23m 46s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3c1d36fe9274133a6db7ef101eeb963 : SUCCESS in 17m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7c997ca29e546e0b2d06360513c1e46 : SUCCESS in 21m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/009b210735a2489fb5352bdfea586625 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"18cbce86dbab50c7b0f1cbb1b4315230bded96a2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-14 15:11:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c516b13746029712c641c34e8167a26bd96f3e6d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-15 01:12:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a160be558be3954f91605d2a938c8feaa6704212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 01:13:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41f6814aa7b01ec93e831f4118ee61aa8e5728c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 02:54:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20e42bf0105b4424af013eca1bb3503b : SUCCESS in 11m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28c6ee9154b640b7a459da3fe6f9fd94 : SUCCESS in 11m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e159e35ce55404faa6475849dc376ed : SUCCESS in 10m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/36f2a6e09a6240c88708fca81bdc244c : SUCCESS in 12m 26s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ce7c7b09a934077a7072de71f0c2738 : SUCCESS in 11m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf890bf46de14c9eb7a4fb7948441e02 : SUCCESS in 14m 16s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a102c1ef85c47bdbc8e502b2b1e226b : FAILURE in 9m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9e411334f864c2ab6ff2f9106a06fe4 : SUCCESS in 1h 36m 48s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1865c66d709a4e68af42e5aa299ecbac : SUCCESS in 21m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17e8c8749a434969a4b5f6991778bcb4 : SUCCESS in 27m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2b6197303bd42c780757c123e645bca : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ee970ca6eeb368c49a3306a8f46aa0586af51fcc","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-15 02:59:03.000000000","message":"Patch Set 1:\n\nreverify","accounts_in_message":[],"_revision_number":1},{"id":"ca64c7fb4f015b8c2b501e2933fb5b537be15096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 04:24:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e57aba50e8d43c9aa10a70623274c32 : FAILURE in 15m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e607ff8c25e147688c1f65d7f76fd3c1 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8344a75841974d8b8db7d60a02401566 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9010094f8c56443981054ef6ec8e30f1 : SUCCESS in 12m 02s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e318a0d2971e467892c80574199c2408 : SUCCESS in 10m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f926806012454976ab57b5092f0845bf : SUCCESS in 12m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b88e0ef74eb74ff58df3787dcb91ae10 : SUCCESS in 8m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b3a33fa2dcd46788f1b7bd4bcd7c8a0 : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad9034f9707147629eb6ac36b0269e5e : FAILURE in 28m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb183fb8c5f940e4ae82f5d0b2fef9c2 : SUCCESS in 1h 24m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4440500d5014c79a81e0af6cecd029e : SUCCESS in 17m 39s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bda5e7c6d554c85b69f6d7aefa82750 : SUCCESS in 20m 30s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33e23b3aa5d4482a850c2155e0312ebc : SUCCESS in 14m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a92c2c88d9b54686b4fc1318018155f7 : SUCCESS in 15m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1f1172e3cef4eb3a93c7c5735db8d64 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5b01ebb0bdb8003abb1729dcb8dab47b86d066f4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-15 05:52:52.000000000","message":"Patch Set 1:\n\nreverify","accounts_in_message":[],"_revision_number":1},{"id":"91e55d085c8b69eac6857db22f5295c37e660a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:13:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/465046d70a2c4882a819181315849991 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09a362d2eda046d1b2ae8ddf92acd174 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6e9fae7f7c649f8b87cbfe710e2a80d : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/57e7c544033f47a3bf6e8e07e48d5903 : SUCCESS in 11m 25s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e6a0ca973554883bceb5e03685f7646 : SUCCESS in 11m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cd87c76615c4d61998b4977e346fba6 : SUCCESS in 10m 01s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed26d8aeae21413eb35942e1ba04bf91 : SUCCESS in 9m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dcaf62c62cf471298e7c40cd56ad2c0 : SUCCESS in 8m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09da27b000b74cc39f8b90e7a9969fa4 : FAILURE in 31m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/763cfcad649045a19b73044dedbd2939 : SUCCESS in 1h 18m 56s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a980e3a483a54abca89fdcacb0a60211 : SUCCESS in 17m 49s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bda5e40912f74491b77e9c3f32df3b57 : SUCCESS in 22m 40s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c4383f07e34427c96a09d4114b92dbc : SUCCESS in 16m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/657daf49c3cc46d89f522e461cfde622 : SUCCESS in 17m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f1af8963cb848d59be28995fca00b7e : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ae0e519808f24829110d861496647cf2517873f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:14:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b59217ac27a52b5ce128fec8f0c472fa3b9c34f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:48:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0d5cd47700e4f6fa9503da3a8a2f497 : SUCCESS in 6m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f368736f63574fc682adf22fac6456d1 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cb7ee2abb79463596ba107cf7d6bb9a : SUCCESS in 5m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3ff14208eb14f89b7699c40623d0ed3 : SUCCESS in 18m 58s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/86678a0e480d49a28e870e27c58bf656 : SUCCESS in 16m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03453b7d2367473d9c9ec86fd8708bc6 : SUCCESS in 13m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/216a30746cd54f2e846a232ca9c0952f : SUCCESS in 17m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/accd543ad5b641d598a955215fc58077 : SUCCESS in 1h 34m 22s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/69955aefdbc54767a1fd0d00501683d9 : SUCCESS in 27m 18s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4e313fe0da943368122b5b6d2750fc3 : SUCCESS in 18m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a28dbc0542c249b8a4f14894f3126aa7 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8a9757d8169a512885d55306dae0b9fe0d4c4dbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:48:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee64017d9c8a2a55585704fd354428a95e5b5458","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:50:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1f966118ab8430bb020a2e546f4478b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de3ab7883f3786b223392b1aaaf3cb52c08cbb8f","revisions":{"de3ab7883f3786b223392b1aaaf3cb52c08cbb8f":{"kind":"REWORK","_number":1,"created":"2021-09-14 08:02:55.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/87/808887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/87/808887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/87/808887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/87/808887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/87/808887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/87/808887/1"}}},"commit":{"parents":[{"commit":"edce396aec8de93b580af21e6b1698dcd582b862","subject":"Merge \"Allow purging policy files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/edce396aec8de93b580af21e6b1698dcd582b862"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-09-14 08:02:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-09-14 08:02:09.000000000","tz":480},"subject":"Add watch_log_file option","message":"Add watch_log_file option\n\nAdd support for Using logging handler designed to watch file system.\n\nChange-Id: I531c77be952daf3f0e86185b9588307da2d1e492\nCloses-Bug: #1943212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/de3ab7883f3786b223392b1aaaf3cb52c08cbb8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/de3ab7883f3786b223392b1aaaf3cb52c08cbb8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
