)]}'
{"id":"openstack%2Fpuppet-mistral~848348","triplet_id":"openstack%2Fpuppet-mistral~master~Ic8154cb670c821ccf06c33721ac73f89c25d78c8","project":"openstack/puppet-mistral","branch":"master","topic":"apache-headers","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-30 13:52:58.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ic8154cb670c821ccf06c33721ac73f89c25d78c8","subject":"Expose headers option of apache::vhost","status":"MERGED","created":"2022-07-01 00:58:09.000000000","updated":"2022-07-30 14:54:22.000000000","submitted":"2022-07-30 14:53:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848348-apache-headers","meta_rev_id":"c7edd37eb0196233f5c6b99e42916f7954cead31","_number":848348,"virtual_id_number":848348,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-30 14:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-07-30 13:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-07-06 14:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-07-30 13:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-01 02:18:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-06 14:21:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"8000fb10ee29ca317cff3f22465024f547f22da9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-01 00:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9226045118ca4f1647545254e0da95371a2b8643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-01 01:27:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"880e7e8cc02e2c30eb3f1d748d2d8bb5d7a6b39c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 02:18:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/921197732cbc4171a767080797760fd7 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2177d1d404249c6899489ea4bab3132 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/154b35ee286e4e0d93cd88bc38b1572c : SUCCESS in 2m 54s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6af5dcb6cf6e4ae8873cfd806c53b251 : SUCCESS in 9m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55df70c05df24d518cd17d5e7f694a9c : SUCCESS in 7m 03s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cbcbd16c3bb4d02ae10aa70a629d19d : SUCCESS in 7m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be9242f699764c858ca71696f8346a6e : SUCCESS in 7m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dde8b4a41ee4c549e8ae318e6fdc05e : FAILURE in 29m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad793e9a8d7a43f09138f51a57eb9286 : SUCCESS in 47m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cea2805a282431d9c06d96cb4a3f2dc : SUCCESS in 12m 53s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/57e3e4e96d304867ba88ee62bdd95cef : SUCCESS in 16m 45s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/810d4a0a89004ebaa1731f4002bb4bf2 : SUCCESS in 14m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb147f75e9844ec79ff6d773eca30156 : SUCCESS in 12m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2129ff3b59f049b0b958b2ac461e0540 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"5e16135b55d664807d684000432d6d7087542204","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-06 14:21:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f1d8421da08c50f1c56384c9cb45ab14598a6862","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-30 13:52:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1c2fcb506bdbfeb547466c077dc45427403da1cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 13:54:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e6d3d527ace194b9496c82620a9d7d298bc9f922","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 14:53:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab28434e3054499a9965cd6fb2b7578\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64cf1f4231ab467c81131af06adf9d31 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b1e866371954ec591470eaaa8eddcfd : SUCCESS in 4m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f93a37462c20443d91ddb117c23cc4fd : SUCCESS in 3m 13s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/772b1cd1d7be4641b5d0ba1c6ff366e1 : SUCCESS in 13m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fd1e2bf3f2e4bc19b0e6e63dc85d6c3 : SUCCESS in 10m 10s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6415ea47600347ee93ce18c63e4d34b5 : SUCCESS in 9m 29s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4beffaf413c457a89511b74a1ad894d : SUCCESS in 58m 32s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3aba77dda35243e3817ee345b0b59ec3 : SUCCESS in 27m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d34276432eed4fd99010d410425ab93d : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"45a2072f6216daf7e6041cd96c1fb93409c26e87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 14:53:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c7edd37eb0196233f5c6b99e42916f7954cead31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 14:54:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f75c90eb73ab4d96aaca53f76a953d75\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ac7255a990440a4917e898fb16f9e38 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf0f7ec1729eb409e550e0e57a3e487aa37f611e","revisions":{"5b059bd00714ad60bbf514a2651fb42d8d0d6aae":{"kind":"REWORK","_number":1,"created":"2022-07-01 00:58:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/848348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/48/848348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/48/848348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/48/848348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/48/848348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/48/848348/1"}}},"commit":{"parents":[{"commit":"acb2f110940e2f3774a9f9ae71971af24e4a6eb9","subject":"Fix typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/acb2f110940e2f3774a9f9ae71971af24e4a6eb9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-01 00:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-01 00:57:59.000000000","tz":540},"subject":"Expose headers option of apache::vhost","message":"Expose headers option of apache::vhost\n\nThe headers option in apache::vhost is required in some case, for\nexample when adding the X-XSS-Protection header. This change allows\ncustomizing the option for the api vhost.\n\nThis change also adds support for request_headers so that both request\nheaders and response headers can customized.\n\nChange-Id: Ic8154cb670c821ccf06c33721ac73f89c25d78c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5b059bd00714ad60bbf514a2651fb42d8d0d6aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5b059bd00714ad60bbf514a2651fb42d8d0d6aae"}]},"branch":"refs/heads/master"},"bf0f7ec1729eb409e550e0e57a3e487aa37f611e":{"kind":"REWORK","_number":2,"created":"2022-07-01 01:27:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/848348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/48/848348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/48/848348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/48/848348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/48/848348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/48/848348/2"}}},"commit":{"parents":[{"commit":"acb2f110940e2f3774a9f9ae71971af24e4a6eb9","subject":"Fix typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/acb2f110940e2f3774a9f9ae71971af24e4a6eb9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-01 00:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-01 01:27:20.000000000","tz":540},"subject":"Expose headers option of apache::vhost","message":"Expose headers option of apache::vhost\n\nThe headers option in apache::vhost is required in some case, for\nexample when adding the X-XSS-Protection header. This change allows\ncustomizing the option for the api vhost.\n\nThis change also adds support for request_headers so that both request\nheaders and response headers can customized.\n\nChange-Id: Ic8154cb670c821ccf06c33721ac73f89c25d78c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/bf0f7ec1729eb409e550e0e57a3e487aa37f611e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/bf0f7ec1729eb409e550e0e57a3e487aa37f611e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
