)]}'
{"id":"openstack%2Fpuppet-mistral~852187","triplet_id":"openstack%2Fpuppet-mistral~master~Idde1ac828eb1be5dacf2f9e6d050aac097c5e910","project":"openstack/puppet-mistral","branch":"master","topic":"cleanup-coordination","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-12 07:51:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idde1ac828eb1be5dacf2f9e6d050aac097c5e910","subject":"Remove deprecated mistral::coordination_* parameters","status":"MERGED","created":"2022-08-04 15:59:34.000000000","updated":"2022-08-12 07:52:47.000000000","submitted":"2022-08-12 07:51:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852187-cleanup-coordination","meta_rev_id":"17baf0566f11e05d9132df692900a8ff1620888d","_number":852187,"virtual_id_number":852187,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 07:51:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-08-08 15:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-12 06:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-12 06:41:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 17:00:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 15:19:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-12 06:41:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0b55d5ad6ade63d3a2a6b6e912eb3dcdb130bb53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 15:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee9bd7591fa0b1ec5858ce39fe1e955d6026880a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:12:31.000000000","message":"Topic set to cleanup-coordination","accounts_in_message":[],"_revision_number":1},{"id":"5852f5f3ff9ecb7537b30044f0bf6b8e8570abb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 17:00:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d051c077ecba4f1685d57ff8e6f242cf\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8351be94e4df4fa4a2a8b7c0da7be65c : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dccb32bf4a054b5c9b33a8293c8ea41f : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3adcb38fef2a46678c796dfc174de58a : SUCCESS in 4m 04s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7dd58607c0a648f2a9ba8faf101517f4 : SUCCESS in 14m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28f0d500c6fd473abf564e1da9455fa3 : SUCCESS in 11m 04s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3a9c98530a14a8480f9cfd57cb55523 : SUCCESS in 10m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09b66e4627eb4c47a50a54726d6eb85f : SUCCESS in 10m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91f31270d3944e3dbc39692c6adfc194 : FAILURE in 36m 24s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95236b175891462b970861a67675118d : SUCCESS in 56m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbe1aee0fb804bce82de9062d571bec1 : FAILURE in 18m 46s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/406296f15b664c62a0dfc63b9fe8bb52 : SUCCESS in 23m 25s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e314349fcb9e49e380e00194f498af37 : FAILURE in 18m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5db8fa2cf3d4bfba05d87099409d426 : FAILURE in 19m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc6c528a9ce4b02b513d0289c2a9afd : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d8ca1e1e8b1e854a91b30bd619d3896d9812cfe5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-08 15:19:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5ed8e7a93e4b956d1a273d43f9dd3a6850d25d3f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-12 06:41:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e000a95fa569d424085a029c938489973c3aeb62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:42:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a922a2020c457abe5df2c232f43805483ca4380b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 07:51:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca5582730c294e30b36cf64cb47015fd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b094d9316b4499ea8385eb714473cea : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8af3a5aaf866486f94e9a36544bfd186 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89b46294205443e19eb793642891f8f1 : SUCCESS in 3m 50s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcb65a8dbd7544a694fe6597165af707 : SUCCESS in 17m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c88a9acf55ca45bab2233460c48cc659 : SUCCESS in 11m 34s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f21c9e281577461797bc27cb6e21d03f : SUCCESS in 11m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d574e913e4b4452194fc9c39d28121c2 : SUCCESS in 1h 07m 29s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/532d68f96d1446c6b5e857838de8d6ca : SUCCESS in 23m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9324514fbd074daca056a4395674ad4b : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a0eac3fd6dd8570bf75cbd5d94a214c922f40fbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 07:51:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"17baf0566f11e05d9132df692900a8ff1620888d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 07:52:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f03834243b4eb5a13be80dd9e545a4\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae766e139cd141589fe9e252ad992402 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6810cd771ad24449c41ffa3e5fca176e41957d5","revisions":{"a6810cd771ad24449c41ffa3e5fca176e41957d5":{"kind":"REWORK","_number":1,"created":"2022-08-04 15:59:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/852187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/87/852187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/87/852187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/87/852187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/87/852187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/87/852187/1"}}},"commit":{"parents":[{"commit":"a29027b27f5770a947df9abbcfaa9485b35b54a7","subject":"Merge \"Remove deprecated mistral::engine parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/a29027b27f5770a947df9abbcfaa9485b35b54a7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 15:57:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 15:57:52.000000000","tz":540},"subject":"Remove deprecated mistral::coordination_* parameters","message":"Remove deprecated mistral::coordination_* parameters\n\n... because these were deprecated during Xena cycle[1] favor of\nthe independent mistral::coordination class.\n\n[1] 298c6ec174780d97966d1724107910ab88d26def\n\nChange-Id: Idde1ac828eb1be5dacf2f9e6d050aac097c5e910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/a6810cd771ad24449c41ffa3e5fca176e41957d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/a6810cd771ad24449c41ffa3e5fca176e41957d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
