)]}'
{"id":"openstack%2Fpuppet-mistral~871429","triplet_id":"openstack%2Fpuppet-mistral~master~I76fa6b53328cbb329a8d5e1fc3ffb00674f2a7d1","project":"openstack/puppet-mistral","branch":"master","topic":"policy_default_rule","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-24 08:54:13.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I76fa6b53328cbb329a8d5e1fc3ffb00674f2a7d1","subject":"Expose policy_default_rule","status":"MERGED","created":"2023-01-23 05:37:06.000000000","updated":"2023-01-24 11:15:50.000000000","submitted":"2023-01-24 11:14:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871429-policy_default_rule","meta_rev_id":"a87e8f58c6e70af2f90a817906f6e384f72491cb","_number":871429,"virtual_id_number":871429,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 11:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-01-23 17:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-01-23 19:31:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-01-23 19:31:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 06:40:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 17:57:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-01-23 19:31:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"ac8a5531ec3e8607bbc76298737444155a895706","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 05:37:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ba66b79d7a39f50254256852fea0e9df5f8143a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 05:37:30.000000000","message":"Topic set to policy_default_rule","accounts_in_message":[],"_revision_number":1},{"id":"03a0fcf81e7c3c2ec2ec1bd59ccddece0d2d4555","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 06:40:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2f9541e5e4a4f518c12d7017e4df694\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f99f32a99974487884164606e34a96a6 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3a6f30e66ac4135a2811f9d71698a8c : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/059e87eb85124fb993b97bff9fd19fbe : SUCCESS in 10m 06s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ea007f768f145f28ef6fc20805b7897 : SUCCESS in 7m 00s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba5328a3ddac4488bb5ceb50a3352c50 : SUCCESS in 7m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51110b33272e4ce8887f47aae766f8d4 : FAILURE in 26m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f160529656e46aebaac3d8bae8185e3 : FAILURE in 1h 01m 49s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2dc19e7403de4951b9f893ed4ec62bbe : SUCCESS in 20m 27s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fc00b89d2534ed7963458bd35884d32 : FAILURE in 6m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2bd05ca36f9477b983a0aa5fe250aeb : FAILURE in 5m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24d56318cb464fa2b2b4797b37d19b0e : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"27762c7fb4ce46a11c379e9c2db3d9ca3c6b278d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 06:50:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f199d820da27a05215e1ee6454918da6efe8a7da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 07:47:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b600e43cc8df40c79e29847cf7e6dcf2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a217ea70c1444f80a2335dfcab388ae8 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2a5a49b0f344947a81061bece26e51f : SUCCESS in 3m 07s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b503fae32b734a819dab11e5d98fc181 : SUCCESS in 11m 28s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/625fd31fc5e84c92a28384b2e7cbd9e8 : SUCCESS in 8m 02s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59c617214bf64583bfd1abc16cca92ba : SUCCESS in 8m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0acfff09646743acb182b8d045387996 : FAILURE in 27m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afa94fc108c6460b9add75ee43bbc9f9 : SUCCESS in 54m 25s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/00c760affb1046c5b1b0641a44fa1198 : SUCCESS in 19m 48s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a981a9e5f6084eddb84cb2d1edce8d6b : FAILURE in 8m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/461b8b5abd1c4047a774c6b84cc34c18 : FAILURE in 8m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/447ff6f232f44816be8d147eb68b1901 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ca2c9a8751f5f5696520aa403549200ea852dccd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-23 17:57:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"292a1e2be3b781fe429adcd854ad91651a03f45e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-23 19:31:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d48aebbce51632094e0cf646bba6f4566c48ef51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:32:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"950ab0d2cddedc92cef87fd26ea697e1f5b2f5a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 20:35:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e148a0d6b8a9454ba0c9235c21e54e2c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38c2d05f43a64a71b62ede5292b69036 : FAILURE in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba016b1303214a3dafdf2b799d4d14f5 : FAILURE in 2m 37s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/681b0df5e1034984b921529d1fefc7dc : FAILURE in 8m 11s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d30e6b41e9b48f1bc35a1b5dfc3d884 : FAILURE in 7m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5158a2391d5490da24fee10fb19004f : SUCCESS in 58m 58s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/809437a4b160429ba5a1f8231976f081 : FAILURE in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f96371e21fb448d08a1c7b2dc369f369 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1b2336b0bd3d350aa0ac0c0c2856293c30cf0f23","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-24 08:54:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69942ceecede1ca740f86930cd5cd371df8b602a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:09:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8cc5eb1f44b4c3a83ca9411837bf859\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fab5b06b19b46b4ada0286730ca358a : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72cf1172c18943d0b056247435c728a8 : SUCCESS in 2m 41s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d963299b0d34284b3890a43b937c7ec : SUCCESS in 10m 51s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84d3965bc382413dbbb82fe69d810143 : SUCCESS in 6m 39s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ec6b14a792d454da690d3c72cd64462 : SUCCESS in 6m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db6763ddaa784be88386c9ac063578b8 : FAILURE in 34m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0199d52e9d334fbb8f302b27178c34d5 : SUCCESS in 1h 09m 27s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/78b2ef85eb2e429d9d6b95ae044e908f : SUCCESS in 21m 00s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3c76b289ab0462e88a91e20b06e0639 : SUCCESS in 15m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29e271a2068f4c51a413a8a48225cfee : SUCCESS in 13m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c37324af52b34fc4911a1785b07e07b2 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"31cd19cdad0ed0754e5b8c073664de34cf032c05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:10:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95ae3c9774864903022f38f02d9286315ffd32fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 11:14:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/964f79f7a53b40fda6ce6003c0473735\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/754e891281614d29ac3b439e8f10297e : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cc5a34e12f74f65b95c1e10e726e90b : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/781133036e0c4e77a442be82b095890c : SUCCESS in 9m 17s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc77afea4ebf4d9e8fd53db460f27eaa : SUCCESS in 7m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ccb935670394ffa9215b848a8945b56 : SUCCESS in 1h 00m 09s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa18aa6576914f7798881dd60eff04a2 : SUCCESS in 19m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36fe77a3790b4ff9b2137e9ba6002b57 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6f42933373de7b92cfc3470d17c8413a29e04d56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 11:14:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a87e8f58c6e70af2f90a817906f6e384f72491cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 11:15:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0412b68e7748699d6c4383e6c3d4f7\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65e668a5e2824607aecb9fd1bb537d73 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3acc248af133a5dec45e31e21851e60663e2112","revisions":{"a3acc248af133a5dec45e31e21851e60663e2112":{"kind":"REWORK","_number":1,"created":"2023-01-23 05:37:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/871429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/29/871429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/29/871429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/29/871429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/29/871429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/29/871429/1"}}},"commit":{"parents":[{"commit":"d15fb87491efd8f9ae3df561c774529c8ce9530d","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/d15fb87491efd8f9ae3df561c774529c8ce9530d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 05:35:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 05:36:53.000000000","tz":540},"subject":"Expose policy_default_rule","message":"Expose policy_default_rule\n\nThe option has been managed by the underlying puppet-oslo module but\nhas not been configurable. This introduces the parameter to customize\nthe option.\n\nChange-Id: I76fa6b53328cbb329a8d5e1fc3ffb00674f2a7d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/a3acc248af133a5dec45e31e21851e60663e2112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/a3acc248af133a5dec45e31e21851e60663e2112"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
