)]}'
{"id":"openstack%2Fpuppet-mistral~902791","triplet_id":"openstack%2Fpuppet-mistral~master~I15ce5ed4f607b0df9b386bb67326db3f790a3d37","project":"openstack/puppet-mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-08 02:58:44.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2023-12-08 04:13:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15ce5ed4f607b0df9b386bb67326db3f790a3d37","subject":"Fix wrong wsgi script path","status":"MERGED","created":"2023-12-06 16:18:33.000000000","updated":"2023-12-08 04:13:36.000000000","submitted":"2023-12-08 04:13:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902791","meta_rev_id":"5f9ca3de87abbb28e9137e0d2d9b8baf4723d41b","_number":902791,"virtual_id_number":902791,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 04:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-12-08 02:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-12-08 02:58:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-06 16:59:48.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-12-06 18:45:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-08 02:58:44.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"8e67c27dedfc7e03971df399ea67be93b3717dbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-06 16:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fbfa1834256e0e15dcdaa2830266f312e729b2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-06 16:23:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c15ac8067a35f66a34a0b1bb9026b90d19f2c8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-06 16:55:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cc4141017355076b2885520b7862e712e85d93ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-06 16:59:48.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8467d51c7125828353e4a8b03578794444df93c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-06 17:36:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0f8b5439c02497fb50106bf65b102141143eebc7","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-06 17:36:57.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7019f15704830f61afd946572de11fff82fb96fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 18:45:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/089661448220418d926a3ca3e1d0037d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b879c11a10c4e98b144ac5b4f8a78fa : SUCCESS in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb2ed367e6e04018b6cb62b9abc52c7a : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/91caf338b81a4df5b05d3cab394db408 : SUCCESS in 6m 02s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d306557eb6e4cddacd1ba005fdce701 : SUCCESS in 9m 33s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e17cc5a95a64609897d60f8fdca7635 : SUCCESS in 7m 21s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/804ca86680584b1ea3cd81322aeb665e : SUCCESS in 10m 44s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c2611cc29bb40f29011e2d9634ce5e3 : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a01f4000a894e50a52037db670a3d85 : FAILURE in 52m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1a28693aef644b4ae8aaf0c852d5ba5 : FAILURE in 1h 07m 50s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/befb282dc9784a2d9afbb0a906c35b64 : FAILURE in 31m 24s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d26779b9c264bbbb3f96b43c4db12f5 : FAILURE in 26m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6db36dca92934168b8885725ddff25d8 : FAILURE in 25m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"078a4684f2764289e1d66f74540ff0e6c2414656","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-07 08:14:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36982ee05afc8ede61da6f004dd4ff55754212b8","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-07 08:34:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"201cf019214fbee685c307e8ce1f3245893345d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-07 08:35:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"92383b4bb14529c990bd4661111e50d3c9bcafd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 08:35:27.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/puppet-mistral on branch master.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/puppet-mistral/.zuul.yaml@master\", line 6, column 17","accounts_in_message":[],"_revision_number":5},{"id":"9304f09aba05e2ec40ef97cf990f3b46bcc4afae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-07 08:42:10.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cf7e469ce49940160581655a336ee0a807a5df6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 09:49:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cfc67ef49774ee8a112924b7a7a179a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9ec51fb80e84b7b8e7063f9b593eb4a : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32b8b6490d7447b084aff282ee767246 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4eb3770a34a41378286716570f39642 : SUCCESS in 5m 49s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/3305f2d9e812471b86387414ff9070c3 : SUCCESS in 10m 51s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a1be61e72be4cdf96f8a7dfcb26a9df : SUCCESS in 6m 53s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc27b43b204c4d2ca4757f87d4c07812 : SUCCESS in 9m 19s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59d968d011cb44528d6a9a22d2657689 : SUCCESS in 7m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57a80dd20ec1470b81faa3d322aa5880 : FAILURE in 58m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3e845f965984fdcaed15f0c7499b65f : FAILURE in 1h 02m 12s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3667651ffd645e2a8b04f8c570c6414 : FAILURE in 30m 32s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b224226e4f2e4c66a84942c9098844ce : FAILURE in 23m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0993719cfcf473ebb46a588dd0000d6 : FAILURE in 24m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5d382a177375d7b8bcfd4f7819ca224fef531151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-12-07 10:49:52.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"e4e43b268318271ef0592749e4389806034f7305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 12:11:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ce86626f7da40bcb573830e8a0ce2dc\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b97d34f8e7b4183b34757f23aa96317 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94c9ef315cb043a4aec8c10b65e647cb : SUCCESS in 2m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/87e98781e894435a9e12139d6cc42537 : SUCCESS in 5m 14s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/30f1fe39b2e54eeb82fd36f2a749a0c3 : SUCCESS in 11m 58s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74e98dd5f2a043f58077752adaf4d752 : SUCCESS in 7m 20s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f5e3bf6cc4541cdae5d4cfd950fe78c : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3364fd24b48483897c097abf40017b4 : SUCCESS in 8m 58s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77320cf2dc3f40bea242919820d4f7c5 : SUCCESS in 48m 47s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec35a94c6a844bdfb4bb251fb1f52149 : SUCCESS in 1h 03m 31s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/caa0a9aa269c4b41b45ba1d5cafab728 : SUCCESS in 22m 27s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3da4e0445ee41ab923736050226caab : SUCCESS in 15m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a9d5f4f54ce4eb6916f06d38259237b : SUCCESS in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6f5e1722793680920559b636a0453f50bcbf8665","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-08 02:58:44.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"61896bfe9c281780c8e6cbde1eeefc1751cc59e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 02:59:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f8cda897a8fd2cb81f3bd8440e35d378b5828bec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 04:13:36.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d733dc079c9a4b24bacbd07b25b68d26\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6404f86933644508161ee67d35dbb74 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0088cae71e3427bbde2320e297d0676 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/97c74285ff4746ff911b7a0b7d403f01 : SUCCESS in 10m 38s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8ad3455a6ff423e8cd01717b08fa900 : SUCCESS in 8m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4e5f13aa38d4396bc33fce0166e7a86 : SUCCESS in 1h 05m 09s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf67abe3fa084ae2be13dd61e60a6a34 : SUCCESS in 20m 58s","accounts_in_message":[],"_revision_number":7},{"id":"5f9ca3de87abbb28e9137e0d2d9b8baf4723d41b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 04:13:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8588e30b3953f377ca5700fa8641824df7166f9c","revisions":{"5771a65f9558dcaa415cb4ecc7105f9fb386c056":{"kind":"REWORK","_number":1,"created":"2023-12-06 16:18:33.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/91/902791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/91/902791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/1"}}},"commit":{"parents":[{"commit":"60d5294c9b0b431c3cb20adb2e55d7844e3bf349","subject":"Merge \"Use new openstackclient tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/60d5294c9b0b431c3cb20adb2e55d7844e3bf349"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:16:27.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:18:32.000000000","tz":60},"subject":"Fix wrong wsgi script path","message":"Fix wrong wsgi script path\n\nSince [1] the wsgi script is not generated with PBR anymore.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/900566\n\nCloses-Bug: #2045782\nChange-Id: I15ce5ed4f607b0df9b386bb67326db3f790a3d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5771a65f9558dcaa415cb4ecc7105f9fb386c056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/5771a65f9558dcaa415cb4ecc7105f9fb386c056"}]},"branch":"refs/heads/master"},"9108f1fbecf50a1ba8dd621d31761cb9b8f3c588":{"kind":"REWORK","_number":2,"created":"2023-12-06 16:23:57.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/91/902791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/91/902791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/2"}}},"commit":{"parents":[{"commit":"60d5294c9b0b431c3cb20adb2e55d7844e3bf349","subject":"Merge \"Use new openstackclient tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/60d5294c9b0b431c3cb20adb2e55d7844e3bf349"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:16:27.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:23:52.000000000","tz":60},"subject":"Fix wrong wsgi script path","message":"Fix wrong wsgi script path\n\nSince [1] the wsgi script is not generated with PBR anymore.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/900566\n\nCloses-Bug: #2045782\nChange-Id: I15ce5ed4f607b0df9b386bb67326db3f790a3d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/9108f1fbecf50a1ba8dd621d31761cb9b8f3c588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/9108f1fbecf50a1ba8dd621d31761cb9b8f3c588"}]},"branch":"refs/heads/master"},"e6ec4bcc5e9ae2124ea256f25ea3fa9959bbe03d":{"kind":"REWORK","_number":3,"created":"2023-12-06 16:55:22.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/91/902791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/91/902791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/3"}}},"commit":{"parents":[{"commit":"60d5294c9b0b431c3cb20adb2e55d7844e3bf349","subject":"Merge \"Use new openstackclient tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/60d5294c9b0b431c3cb20adb2e55d7844e3bf349"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:16:27.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:55:18.000000000","tz":60},"subject":"Fix wrong wsgi script path","message":"Fix wrong wsgi script path\n\nSince [1] the wsgi script is not generated with PBR anymore.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/900566\n\nCloses-Bug: #2045782\nChange-Id: I15ce5ed4f607b0df9b386bb67326db3f790a3d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/e6ec4bcc5e9ae2124ea256f25ea3fa9959bbe03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/e6ec4bcc5e9ae2124ea256f25ea3fa9959bbe03d"}]},"branch":"refs/heads/master"},"e4335face384ac6899b0bff8985b848f68f0e3f6":{"kind":"REWORK","_number":4,"created":"2023-12-06 17:36:37.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/91/902791/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/91/902791/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/4"}}},"commit":{"parents":[{"commit":"60d5294c9b0b431c3cb20adb2e55d7844e3bf349","subject":"Merge \"Use new openstackclient tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/60d5294c9b0b431c3cb20adb2e55d7844e3bf349"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:16:27.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 17:36:28.000000000","tz":60},"subject":"Fix wrong wsgi script path","message":"Fix wrong wsgi script path\n\nSince [1] the wsgi script is not generated with PBR anymore.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/900566\n\nCloses-Bug: #2045782\nChange-Id: I15ce5ed4f607b0df9b386bb67326db3f790a3d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/e4335face384ac6899b0bff8985b848f68f0e3f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/e4335face384ac6899b0bff8985b848f68f0e3f6"}]},"branch":"refs/heads/master"},"044060bccd9e12ef47c278ff465aff7d77a27ae4":{"kind":"REWORK","_number":5,"created":"2023-12-07 08:35:04.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/91/902791/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/91/902791/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/5"}}},"commit":{"parents":[{"commit":"60d5294c9b0b431c3cb20adb2e55d7844e3bf349","subject":"Merge \"Use new openstackclient tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/60d5294c9b0b431c3cb20adb2e55d7844e3bf349"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:16:27.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-07 08:30:42.000000000","tz":60},"subject":"Fix wrong wsgi script path","message":"Fix wrong wsgi script path\n\nSince [1] the wsgi script is not generated with PBR anymore.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/900566\n\nCloses-Bug: #2045782\nChange-Id: I15ce5ed4f607b0df9b386bb67326db3f790a3d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/044060bccd9e12ef47c278ff465aff7d77a27ae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/044060bccd9e12ef47c278ff465aff7d77a27ae4"}]},"branch":"refs/heads/master"},"1324661945e1054d27f5d2d2511eaf2af54c3860":{"kind":"REWORK","_number":6,"created":"2023-12-07 08:42:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/902791/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/91/902791/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/6"}}},"commit":{"parents":[{"commit":"60d5294c9b0b431c3cb20adb2e55d7844e3bf349","subject":"Merge \"Use new openstackclient tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/60d5294c9b0b431c3cb20adb2e55d7844e3bf349"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:16:27.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-07 08:42:10.000000000","tz":0},"subject":"Fix wrong wsgi script path","message":"Fix wrong wsgi script path\n\nSince [1] the wsgi script is not generated with PBR anymore.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/900566\n\nCloses-Bug: #2045782\nChange-Id: I15ce5ed4f607b0df9b386bb67326db3f790a3d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/1324661945e1054d27f5d2d2511eaf2af54c3860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/1324661945e1054d27f5d2d2511eaf2af54c3860"}]},"branch":"refs/heads/master"},"8588e30b3953f377ca5700fa8641824df7166f9c":{"kind":"REWORK","_number":7,"created":"2023-12-07 10:49:52.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/91/902791/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/91/902791/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/91/902791/7"}}},"commit":{"parents":[{"commit":"60d5294c9b0b431c3cb20adb2e55d7844e3bf349","subject":"Merge \"Use new openstackclient tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/60d5294c9b0b431c3cb20adb2e55d7844e3bf349"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-06 16:16:27.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-12-07 10:48:19.000000000","tz":60},"subject":"Fix wrong wsgi script path","message":"Fix wrong wsgi script path\n\nSince [1] the wsgi script is not generated with PBR anymore.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/900566\n\nCloses-Bug: #2045782\nChange-Id: I15ce5ed4f607b0df9b386bb67326db3f790a3d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/8588e30b3953f377ca5700fa8641824df7166f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/8588e30b3953f377ca5700fa8641824df7166f9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
