)]}'
{"id":"openstack%2Fpuppet-mistral~948701","triplet_id":"openstack%2Fpuppet-mistral~master~Id763f60274d965cf697873b1649b7de27094d842","project":"openstack/puppet-mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-06 12:44:11.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-06 11:37:43.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"Id763f60274d965cf697873b1649b7de27094d842","subject":"Convert (allow|deny)list to comma-separated string","status":"MERGED","created":"2025-05-02 13:54:00.000000000","updated":"2025-05-06 12:44:11.000000000","submitted":"2025-05-06 12:44:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948701","meta_rev_id":"eae04ed76aca7980895337532512456254c686b1","_number":948701,"virtual_id_number":948701,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 12:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-05-05 15:01:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-05-06 11:37:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-05-06 11:37:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-02 14:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-05 15:01:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-05-06 09:53:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"94a54bd226a844af63ba2b51eb70ab5ad93eec67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-02 13:54:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5288bd36b4ffb2c6970a148adf2d590e857bd638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 14:47:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b596d70b5e504e5aa64f0134419d6f68\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c2bd56b558e40c3b1799b9e559c8701 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/808eaadb7e974a6d8eef75ecd7130fbd : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d06d2b5c01d342af92c7a0b4bccb7c47 : SUCCESS in 2m 22s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/815f6ab9ddb243abb16975a621818063 : SUCCESS in 6m 05s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54e9dbeb729e4a869a7c73c52fba7aaa : SUCCESS in 3m 56s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2695319e69de46608fb180c41f71bb8d : SUCCESS in 4m 59s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbdcb92ea9e24c1e92d9347756be4e00 : SUCCESS in 48m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa9d5c886ad442ff8464cd1d32a796e5 : SUCCESS in 49m 15s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/69efb40722d84470b2805bd59f71e885 : SUCCESS in 18m 34s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62d3c74847cd40918bf8208461a00881 : SUCCESS in 12m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/919c22d8dd5242678a64733e2db07ab2 : SUCCESS in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d386adaf338268de0fc47fec2558e9e43e1b397c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-05-05 15:01:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64f818df6992aacc2a36b077b352ec8a2f1e7956","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-06 11:37:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7282a916b3bbb081e8c1ac5ef896ab8061f0b29a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 11:38:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6555a85921cc2282ce230a6374574105b4dfb41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:44:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653a4b02472a4f0f98ae1d3f1b992b25\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5d1223fa62844daa92de4b37b65eb3b : SUCCESS in 2m 14s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52cd72157a2d4fcda9841cfc8f05fb56 : SUCCESS in 2m 11s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a6d2e500cd64ac1874a00acd86e68e8 : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/983f147fc7ab420589714a6d7c16a312 : SUCCESS in 4m 04s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dd950424ae84692b8b86554e27f562d : SUCCESS in 4m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/39b9d650aa9a444fa81bc5f9b82535dc : SUCCESS in 1h 00m 50s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/bce2582b589e49239c8b485aa4d62738 : SUCCESS in 16m 53s","accounts_in_message":[],"_revision_number":1},{"id":"eae04ed76aca7980895337532512456254c686b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:44:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba711f966c0bc3b29a0f660e712eb1c828ac9be0","revisions":{"ba711f966c0bc3b29a0f660e712eb1c828ac9be0":{"kind":"REWORK","_number":1,"created":"2025-05-02 13:54:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/948701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/01/948701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/01/948701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/01/948701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/01/948701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/01/948701/1"}}},"commit":{"parents":[{"commit":"64543a028f9b4b7f89e997d0e17a2ef10c5acffb","subject":"Add action_providers and legacy_action_providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/64543a028f9b4b7f89e997d0e17a2ef10c5acffb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-02 13:53:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-02 13:53:08.000000000","tz":540},"subject":"Convert (allow|deny)list to comma-separated string","message":"Convert (allow|deny)list to comma-separated string\n\nThis is follow-up of 64543a028f9b4b7f89e997d0e17a2ef10c5acffb .\n\nThese options are defined as ListOpt, so array values should be\nrendered as a comma-seprated strings.\n\nChange-Id: Id763f60274d965cf697873b1649b7de27094d842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/ba711f966c0bc3b29a0f660e712eb1c828ac9be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/ba711f966c0bc3b29a0f660e712eb1c828ac9be0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
