)]}'
{"id":"openstack%2Fpuppet-mistral~948993","triplet_id":"openstack%2Fpuppet-mistral~stable%2F2024.2~I6302ff41f1bad8c01defb915077ee621faa05b60","project":"openstack/puppet-mistral","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-07 07:55:38.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-07 15:04:51.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2025-05-07 14:02:27.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I6302ff41f1bad8c01defb915077ee621faa05b60","subject":"Add action_providers and legacy_action_providers","status":"MERGED","created":"2025-05-07 07:16:09.000000000","updated":"2025-05-07 15:05:39.000000000","submitted":"2025-05-07 15:04:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948993","cherry_pick_of_change":948777,"cherry_pick_of_patch_set":1,"meta_rev_id":"b3f9d352fb8941d43e4f5296a71d94c2f8e8fb2d","_number":948993,"virtual_id_number":948993,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 15:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-05-07 09:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-05-07 14:02:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-07 09:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-05-07 14:02:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 07:18:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-07 07:18:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-05-07 07:18:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-05-07 08:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74223ed6bd63251feb488e96cdef4a39d9d7785d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-07 07:16:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"ec1eee145a6aa76835fea10b68299d9bbf575776","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-07 07:18:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9976d1938398358bfca0c0347dfaa82a5fb6a3a8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-07 07:55:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc2bd63a5d009bcee54102cadb8a23c5888cfc84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 08:19:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a68dbed1358c4404a169421a4424bf10\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c46e48ac7d16472c8cb0df5d227cc887 : FAILURE in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cc35e77950d4af0841d0282a666eae4 : SUCCESS in 2m 09s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/efc55b3126a041878af93b2d15758b2d : SUCCESS in 5m 08s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/bff1e89fb5884f4ba557f03d949affc0 : SUCCESS in 8m 04s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/480236ece0a646c7ba6d9a828089aa2e : SUCCESS in 3m 57s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9ab3744e9f041af85a9c2bdf774f54a : SUCCESS in 5m 50s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a81a7cc7b15434287692c3de354d155 : SUCCESS in 3m 49s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f6c9a2db348479cb614737101d3bd60 : SUCCESS in 35m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9109d185a0bb4feda036bdddd5423223 : SUCCESS in 31m 44s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8456a31d08e47e7b3fb861a919c7628 : SUCCESS in 57m 21s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/20175566a3984dc2bc70ea7d404495bd : SUCCESS in 12m 42s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/023cdd4946564afe8424234008dc5f5f : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fecc256c525a463492518494edbd1bcb : FAILURE in 14m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17801a2ae1af44dba2e08f32aad030a0 : SUCCESS in 5m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef3ac740473941f0955f0a46f04fd446 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f71329f84a24d0c51ed824ea7119dd1f40663fa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-07 09:01:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e1c7c7b7da95e4880978e2f6d2d5a2567440f184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 10:04:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b906edf8dd474aaaa5a6c6f2998cdd4f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0fd5991357c47039671b2fbc61da6c5 : SUCCESS in 2m 15s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b39c7f38469640c3ae13481cdcfb58bb : SUCCESS in 1m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/401bb0fd5a1e42a19720a90982b7c19c : SUCCESS in 2m 39s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ba783e96cf8493a92b2978f0b08667c : SUCCESS in 4m 02s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5806dc6382a043849844655c0505f13a : SUCCESS in 2m 39s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c480720d313c47e08c5ba595e09999ba : SUCCESS in 7m 42s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e82d1fb0cc50467baefafd1b780e07a5 : SUCCESS in 4m 37s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e36eb399d12e4bfe96f391c72c61038d : SUCCESS in 44m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5e95e1ed8d3442d91e8defde6b096bd : SUCCESS in 54m 31s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/186690b22ae04bb5a14129e03289c3e5 : SUCCESS in 57m 23s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8fe4fc478c74c7bb3b3184f63caba6b : SUCCESS in 16m 45s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d7748ebabb947709ac8a1dc4d09bc48 : SUCCESS in 12m 07s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/827e69645c3b4230b9476afc50f0ea79 : FAILURE in 12m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebcb670ebabe4f348269e0c161f69846 : SUCCESS in 11m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f38896febd6c494896aa30048b5319b7 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":2},{"id":"97948c59e6bdc05434309ce082b11a3623138c27","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-05-07 14:02:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e591398cf261a8e51940cb5aa0cd5612b44fc90b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:03:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3a5091442db781c0780b433227d4feb7c4c2ec24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 15:04:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be18abfc23f4db7bb15c39574959c94\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/775424f0841d46f194a4a0af430e5e4f : SUCCESS in 2m 15s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fee473b5c45b4c349cfb2be1fbf4c0e9 : SUCCESS in 2m 13s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c2b0fd25d714e0d9e78aa1ad901f474 : SUCCESS in 7m 39s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8292f6e8a58413c8c49a0b947b14cb7 : SUCCESS in 4m 28s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/23e188011c81484f8c710260ed82e3fb : SUCCESS in 8m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df7ee29200e144f9836157f57d1bf238 : SUCCESS in 57m 26s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae396df452ea4abba381ad112d36837c : SUCCESS in 18m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/459ecc14f49e451b8690d17517c331b8 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f445ff48f6fa9e76bdbdeca6a9f937d51b831893","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 15:04:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b3f9d352fb8941d43e4f5296a71d94c2f8e8fb2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 15:05:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a933b02074d84b8b80f250a5db8fcfc1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef25f2e057bc44819881dff3d5b51c62 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b3d3c11964f5adde2ce308db89677921304b0bc9","revisions":{"05be2274f3422d31ac431d9f3f6750f42e98df3e":{"kind":"REWORK","_number":1,"created":"2025-05-07 07:16:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/93/948993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/93/948993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/93/948993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/93/948993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/93/948993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/93/948993/1"}}},"commit":{"parents":[{"commit":"349215b98755c9636afe1ae9cde3bf843587fee2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/349215b98755c9636afe1ae9cde3bf843587fee2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-25 08:20:14.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-05-07 07:16:09.000000000","tz":0},"subject":"Add action_providers and legacy_action_providers","message":"Add action_providers and legacy_action_providers\n\nThis adds two new classes to configure the config\ngroups action_providers and legacy_action_providers\nin the Mistral configuration.\n\nSome config opts were introduced in [1] and others\nalready existed and has been added as well.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/944015/3\n\nChange-Id: I6302ff41f1bad8c01defb915077ee621faa05b60\n(cherry picked from commit 64543a028f9b4b7f89e997d0e17a2ef10c5acffb)\n(cherry picked from commit 1184caee232c3e885955b7fd8bbebfb7fb0bfcd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/05be2274f3422d31ac431d9f3f6750f42e98df3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/05be2274f3422d31ac431d9f3f6750f42e98df3e"}]},"branch":"refs/heads/stable/2024.2"},"b3d3c11964f5adde2ce308db89677921304b0bc9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-07 09:01:55.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/93/948993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/93/948993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/93/948993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/93/948993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/93/948993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/93/948993/2"}}},"commit":{"parents":[{"commit":"2cace5810927cbefceb48bca03fc2fc959d751b4","subject":"Fix new lint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/2cace5810927cbefceb48bca03fc2fc959d751b4"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-25 08:20:14.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-05-07 09:01:55.000000000","tz":0},"subject":"Add action_providers and legacy_action_providers","message":"Add action_providers and legacy_action_providers\n\nThis adds two new classes to configure the config\ngroups action_providers and legacy_action_providers\nin the Mistral configuration.\n\nSome config opts were introduced in [1] and others\nalready existed and has been added as well.\n\n[1] https://review.opendev.org/c/openstack/mistral/+/944015/3\n\nChange-Id: I6302ff41f1bad8c01defb915077ee621faa05b60\n(cherry picked from commit 64543a028f9b4b7f89e997d0e17a2ef10c5acffb)\n(cherry picked from commit 1184caee232c3e885955b7fd8bbebfb7fb0bfcd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b3d3c11964f5adde2ce308db89677921304b0bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b3d3c11964f5adde2ce308db89677921304b0bc9"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
