)]}'
{"id":"openstack%2Fpuppet-mistral~956342","triplet_id":"openstack%2Fpuppet-mistral~master~I546e0b38c69c8e5514e2bfe22396cc78ecbcf18e","project":"openstack/puppet-mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-04 15:48:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I546e0b38c69c8e5514e2bfe22396cc78ecbcf18e","subject":"Support dedicated mistral-notifier","status":"MERGED","created":"2025-08-01 09:16:08.000000000","updated":"2025-08-04 15:49:33.000000000","submitted":"2025-08-04 15:48:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956342","meta_rev_id":"7449c05a3fa5519c5afa3f19a5f9bf59a4b46c35","_number":956342,"virtual_id_number":956342,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 15:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-08-04 06:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-08-04 14:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-08-04 14:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-01 10:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-04 06:53:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-08-04 14:51:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ad4f45d4be6492b0612822b376bef8468387e93b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-01 09:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80c62fc085b61ecfd45fbc1416e67d839565d7f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 10:27:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/057519ce3b0d491791def3ac0caf4f90\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f0fe01c2118470c958e97b0964df86d : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d74c23b8f4f47909581d66ea74d931c : SUCCESS in 5m 00s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8ace0bbb20b4c84991f709f112af4ff : SUCCESS in 4m 45s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ac2f031dbfd44c1a7bdce22a3dae2f4 : FAILURE in 6m 06s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ff09b4afb1e4123a45ea559334536d2 : FAILURE in 4m 19s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afbda4aa9543497b940909b8b5b65139 : FAILURE in 4m 39s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24f23b385c17428f92cdeefb3b06faed : FAILURE in 29m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f3cba23481d449b8472e7d2ce59019f : SUCCESS in 1h 00m 23s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee75fe4bd7c54ebf8fb9fd638aef9275 : SUCCESS in 18m 28s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8344a3cc83a74c34a9fab9d444a9e86b : SUCCESS in 13m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fea567a8bcc14b5aaa2d3b1a3a97e5e1 : SUCCESS in 12m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88920fe60cb54aa5bb5e3da035bc2105 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1},{"id":"693424e8b9b83f6c580f4f385f1b5687f3b70b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-02 12:56:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7140adb9a46ac4aba9f0e2bba09c83bd61838299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-02 13:02:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b44a43caa0a37b2c9fdb4f1a4250db5109820ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 14:01:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/634ca28bfa8240719dbce9949acd2377\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01ba4cbd6c0a4acbaf8d52e0049b9533 : SUCCESS in 1m 46s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/370a2559f59f4e709cd8645b75cea4b1 : SUCCESS in 1m 47s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8416b57fe9140deb38dc403142f1666 : SUCCESS in 1m 44s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea193c2721b440a08f72e736901f6e7e : FAILURE in 4m 38s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/238d948191f94649803a4b41628bb96c : FAILURE in 4m 49s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4834a29a3934dbfa3bc814e6f46e345 : FAILURE in 4m 32s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56ab2a2ec499403dadd2f40eb6d526de : SUCCESS in 47m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bf781b4e1b14f85b8acfcb49d64169d : SUCCESS in 46m 52s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f3b056291c5407abe0241a851df0aa9 : SUCCESS in 18m 23s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1018a21d99b045ef840a76e17669e029 : SUCCESS in 12m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44e89d004cd241d290a43099befdc0d9 : SUCCESS in 10m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17bc2c7c586a4ecebd27645a37083841 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":3},{"id":"6f35c4c111002066aa6b4856f1fe68e3450ed36e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-02 14:17:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a5d356a68b0f69725a788492af550f69ae7c1dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-02 15:08:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e4272b042d641e397e0301a70aeab57\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35ed22255a1a4a718d76a5d906ed3d3f : SUCCESS in 1m 46s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36f706e05cc2409998b58ec4c82e1e3d : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e64d3db9b3141fca9ac43b81cd6d0df : SUCCESS in 2m 15s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9937c5470b0410889bfa50b6c157792 : SUCCESS in 8m 10s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61cca61e16c2427488ce076f35054fa4 : SUCCESS in 4m 57s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13f2eac55dd148379fba140c87323be3 : SUCCESS in 4m 53s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03c8ca6660604d6ba9cdc0181547645e : SUCCESS in 46m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93809f1191ad470a998ca40db05daa7d : SUCCESS in 28m 37s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/91ed4976f23a4310b1da30f055160bb8 : SUCCESS in 8m 00s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/665e8a1b4d3a45d3bbc4065488cefd20 : SUCCESS in 13m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c76a1bc97744060988f304e811257b4 : SUCCESS in 12m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7b3913ade664e5984fec5eefa046fc0 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":4},{"id":"a5b91e69afe2e016fc93b92f4225d7555cb5bcb1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-04 06:53:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f2901feacc6fbc7a30f6cc03806a0d122d6aad56","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-04 14:51:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"91a457824e9016da8ac6b23c2b4224f24408a902","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 14:51:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"161696abdb505069ed4a1dd9472373d45e2021ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 15:48:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77971580557248e49ebd4a7da8c5654b\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b40a8423aadd4959a36089ce6df7759c : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d8f9b831e264b8a9487638901dd8df2 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bd36260cb824e028bd7d2e6092be4b8 : SUCCESS in 2m 58s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/8163dbd126ae48c193c8a29fa63f9c36 : SUCCESS in 6m 37s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a81e3c4da6af4cd8820a1cc52bd9ef75 : SUCCESS in 4m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fa452a4e18d4ccea1aaef402b4159d4 : SUCCESS in 55m 25s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/eec58174573a440f9f2ad7688acf7b8e : SUCCESS in 18m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47bd11bc2d5f405b90e26598bfd73ca7 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":4},{"id":"a9add27bfa48808e77f3a4b8ee81102c2244f712","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 15:48:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7449c05a3fa5519c5afa3f19a5f9bf59a4b46c35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 15:49:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9745566fff64e83805e9b2161de5fd9\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc6469e9b637464ba0cde21063572916 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2beb127791c6adcb887f2c39c378969990238f23","revisions":{"20acb6d0c8a4d40acf3a634608c3642a34d55681":{"kind":"REWORK","_number":1,"created":"2025-08-01 09:16:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/956342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/42/956342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/1"}}},"commit":{"parents":[{"commit":"6ef3fe37ee67df0ceb1cf4e7190abc8e7fe940fe","subject":"Add queue_manager and stream_fanout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/6ef3fe37ee67df0ceb1cf4e7190abc8e7fe940fe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 08:58:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 09:15:55.000000000","tz":540},"subject":"Support dedicated mistral-notifier","message":"Support dedicated mistral-notifier\n\nmistral by default requires a dedicated notifier service.\n\nChange-Id: I546e0b38c69c8e5514e2bfe22396cc78ecbcf18e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/20acb6d0c8a4d40acf3a634608c3642a34d55681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/20acb6d0c8a4d40acf3a634608c3642a34d55681"}]},"branch":"refs/heads/master"},"af0af31cab5f5c49b562dcb3e9a4afb3748451e0":{"kind":"REWORK","_number":2,"created":"2025-08-02 12:56:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/956342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/42/956342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/2"}}},"commit":{"parents":[{"commit":"6ef3fe37ee67df0ceb1cf4e7190abc8e7fe940fe","subject":"Add queue_manager and stream_fanout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/6ef3fe37ee67df0ceb1cf4e7190abc8e7fe940fe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 08:58:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 12:56:27.000000000","tz":540},"subject":"Support dedicated mistral-notifier","message":"Support dedicated mistral-notifier\n\nmistral by default requires a dedicated notifier service.\n\nChange-Id: I546e0b38c69c8e5514e2bfe22396cc78ecbcf18e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/af0af31cab5f5c49b562dcb3e9a4afb3748451e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/af0af31cab5f5c49b562dcb3e9a4afb3748451e0"}]},"branch":"refs/heads/master"},"bdcd743cfeac38e3fb417d67a37e36ac850e1a62":{"kind":"REWORK","_number":3,"created":"2025-08-02 13:02:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/956342/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/42/956342/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/3"}}},"commit":{"parents":[{"commit":"6ef3fe37ee67df0ceb1cf4e7190abc8e7fe940fe","subject":"Add queue_manager and stream_fanout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/6ef3fe37ee67df0ceb1cf4e7190abc8e7fe940fe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 08:58:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 13:02:01.000000000","tz":540},"subject":"Support dedicated mistral-notifier","message":"Support dedicated mistral-notifier\n\nmistral by default requires a dedicated notifier service.\n\nChange-Id: I546e0b38c69c8e5514e2bfe22396cc78ecbcf18e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/bdcd743cfeac38e3fb417d67a37e36ac850e1a62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/bdcd743cfeac38e3fb417d67a37e36ac850e1a62"}]},"branch":"refs/heads/master"},"2beb127791c6adcb887f2c39c378969990238f23":{"kind":"REWORK","_number":4,"created":"2025-08-02 14:17:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/956342/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/42/956342/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/42/956342/4"}}},"commit":{"parents":[{"commit":"6ef3fe37ee67df0ceb1cf4e7190abc8e7fe940fe","subject":"Add queue_manager and stream_fanout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/6ef3fe37ee67df0ceb1cf4e7190abc8e7fe940fe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 08:58:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-02 14:11:27.000000000","tz":540},"subject":"Support dedicated mistral-notifier","message":"Support dedicated mistral-notifier\n\nmistral by default requires a dedicated notifier service.\n\nChange-Id: I546e0b38c69c8e5514e2bfe22396cc78ecbcf18e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/2beb127791c6adcb887f2c39c378969990238f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/2beb127791c6adcb887f2c39c378969990238f23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
