)]}'
{"id":"openstack%2Fpuppet-mistral~962910","triplet_id":"openstack%2Fpuppet-mistral~master~I3e1a7b170c9840b0de83e0eb97c596154790d019","project":"openstack/puppet-mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"37313":{"account":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"last_update":"2025-10-03 14:59:09.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-03 14:59:09.000000000","reason":"Change was submitted"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-10-03 14:59:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e1a7b170c9840b0de83e0eb97c596154790d019","subject":"Add support for mistral_api_uwsgi_config in Debian","status":"MERGED","created":"2025-10-03 09:13:28.000000000","updated":"2025-10-03 15:00:03.000000000","submitted":"2025-10-03 14:59:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962910","meta_rev_id":"8a048d6fa904ee5c54a26174928582ec7bfeb7e1","_number":962910,"virtual_id_number":962910,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-03 14:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":2,"date":"2025-10-03 12:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-10-03 13:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-10-03 13:58:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-03 10:00:38.000000000","updated_by":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"reviewer":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"state":"REVIEWER"},{"updated":"2025-10-03 12:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-03 12:25:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-10-03 13:58:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a9f8a75e36ace1e0c574ef3ab187020ae7357360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-03 09:13:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb8b9e8608ca821931665d178c7da1137ead478b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-03 09:56:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5eb3a925183594fb89769c780bc39ea5f828fd2b","author":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"date":"2025-10-03 10:00:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fa6e25afec9d191cbc4593e4bc108dbb7ba51871","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-03 10:55:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c433d29178ab4a8f74503f44b68a947a068f02e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-03 10:59:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e4e066755c08f536fa402ef09e85c30391dc7527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 12:01:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc87b9aef46249e899d342c9f50c1e8c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cbcfe61dd3a499d99cc9204ea038b2d : SUCCESS in 3m 14s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/070383574e0f4607a1983726667ac485 : SUCCESS in 3m 02s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/eb4fba65d93040589c27e5630dbf3399 : SUCCESS in 3m 50s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5591ef30043d4843846278906b110e98 : SUCCESS in 4m 57s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e32825f619d4704b5c85d6f86e53b84 : SUCCESS in 4m 01s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b64b9c6c22543b785246ec4364e17f1 : SUCCESS in 47m 08s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/259ac6b8a7b3406e857e6c0c2c66e476 : SUCCESS in 35m 45s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/147fcb52e005429c970d8b87da537e4f : SUCCESS in 57m 51s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/33b8127980e2463b87069637a4d5e7be : SUCCESS in 19m 43s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecc6ff75084a4e8d805e837291084099 : SUCCESS in 13m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c18db7fef35b4ceb948618803cc39bf1 : SUCCESS in 12m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90975c6b5a8b44b6a696b765f5718bd3 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"201cac2b03d755e111f3a988094edaa1c846ebe7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 12:25:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b510d6621ad9131ac598347d5a783e6d694932d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-03 12:29:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"de36171ffcd9413dedd1d512917766bb06bae346","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-03 12:29:28.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fddca379d04e1bb37d59b40a189cae031a9b1531","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 12:43:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c1d3eed79a95faaa289d4de6eea1125d865cf9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 12:43:32.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"206bc57c1d44563a64645b0cc5f2ef6e8b76b4f8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 12:43:53.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"36d989db5c01fab9c8835dad6e8fa2e76446e200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 13:46:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58e1ac7fbff24f81a27663fdd3996e02\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ad37bb1f3584a1aa08154b4a7b49286 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c36a54eb0a0b43198c4cb3de8257dbb4 : SUCCESS in 1m 48s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/7ebc0efba25e4495a7155f70a82b3296 : SUCCESS in 4m 47s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c5dd654ed3648f199472cb9e4b3e5ac : SUCCESS in 5m 12s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a705792c6f044eb183a2bdf2b0d0cdeb : SUCCESS in 5m 14s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3488f7138d7543af814521814336da69 : SUCCESS in 26m 44s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4aa31ec821443bc9f46e4e705c8b6c8 : SUCCESS in 44m 39s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b8d45762eb1422f9b34498fc87a2827 : SUCCESS in 57m 36s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e01a584d8fb24ee5ae06f82ebce41d33 : SUCCESS in 10m 41s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72c94982fd184c53964d48e7dcece6f6 : SUCCESS in 6m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dbf8e3e50874cf8a0df3a8513146ebc : SUCCESS in 12m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb2ffa303ca74041b395b7274f8faff2 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":6},{"id":"4c4640145e7b14bd1505555157f03bb41a2e2e5b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-10-03 13:58:03.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ea30c8c8a394ae91a839784f3648937bc476c52e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 13:58:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6b60224ba28d07e03022de726065b827c49ae0f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:59:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc43b1e9e478475aa18d87dda7cbcc8c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cd1639f06744e199dc37ecbf8790e1b : SUCCESS in 2m 31s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/790a480b88c142ceb7c2c0e87999e0aa : SUCCESS in 2m 16s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/d396acf075a64b9aa644dd7471b1526c : SUCCESS in 3m 44s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6818dc053c4d4f39be0b054816e0e8c9 : SUCCESS in 4m 35s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5209dfe69cf4ccbb993cd677ecb7886 : SUCCESS in 56m 17s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a9706548d574388804e938304a3d350 : SUCCESS in 14m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fedc7429249641e482e335f3375a0342 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":6},{"id":"314b54fd33ad48262359a8f62153b0bef6869f5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:59:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8a048d6fa904ee5c54a26174928582ec7bfeb7e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 15:00:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04c741cb8c66444cb575ad610deb9d75\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/607d3f20f0bb4b76a87fede97ff1e07a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"92ca9b40e299829168a10d15219483b7ec9dbe2a","revisions":{"2da20d896b64d9ef9985d6394ae91cbc5a5f1571":{"kind":"REWORK","_number":1,"created":"2025-10-03 09:13:28.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/962910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/10/962910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/1"}}},"commit":{"parents":[{"commit":"05068ff8f60506874193f3d68ec8645ce9fcc717","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/05068ff8f60506874193f3d68ec8645ce9fcc717"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 09:08:19.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 09:13:10.000000000","tz":120},"subject":"Add support for mistral_api_uwsgi_config in Debian","message":"Add support for mistral_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Mistral API. Therefore, this patch adds a new\nmistral_api_uwsgi_config provider as well as a new\nmistral::wsgi::uwsgi class.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I3e1a7b170c9840b0de83e0eb97c596154790d019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/2da20d896b64d9ef9985d6394ae91cbc5a5f1571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/2da20d896b64d9ef9985d6394ae91cbc5a5f1571"}]},"branch":"refs/heads/master"},"50edd1e30ff2a91944d0e5dcb7aab4da2459727b":{"kind":"REWORK","_number":2,"created":"2025-10-03 09:56:26.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/962910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/10/962910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/2"}}},"commit":{"parents":[{"commit":"05068ff8f60506874193f3d68ec8645ce9fcc717","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/05068ff8f60506874193f3d68ec8645ce9fcc717"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 09:08:19.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 09:56:16.000000000","tz":120},"subject":"Add support for mistral_api_uwsgi_config in Debian","message":"Add support for mistral_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Mistral API. Therefore, this patch adds a new\nmistral_api_uwsgi_config provider as well as a new\nmistral::wsgi::uwsgi class.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I3e1a7b170c9840b0de83e0eb97c596154790d019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/50edd1e30ff2a91944d0e5dcb7aab4da2459727b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/50edd1e30ff2a91944d0e5dcb7aab4da2459727b"}]},"branch":"refs/heads/master"},"4bfd61c5764e1f3665d6068bd486f184e14a188b":{"kind":"REWORK","_number":3,"created":"2025-10-03 10:55:45.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/962910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/10/962910/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/3"}}},"commit":{"parents":[{"commit":"05068ff8f60506874193f3d68ec8645ce9fcc717","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/05068ff8f60506874193f3d68ec8645ce9fcc717"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 09:08:19.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 10:55:32.000000000","tz":120},"subject":"Add support for mistral_api_uwsgi_config in Debian","message":"Add support for mistral_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Mistral API. Therefore, this patch adds a new\nmistral_api_uwsgi_config provider as well as a new\nmistral::wsgi::uwsgi class.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I3e1a7b170c9840b0de83e0eb97c596154790d019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/4bfd61c5764e1f3665d6068bd486f184e14a188b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/4bfd61c5764e1f3665d6068bd486f184e14a188b"}]},"branch":"refs/heads/master"},"28118de6f25e6d807f436ffefacd489274e64e24":{"kind":"REWORK","_number":4,"created":"2025-10-03 10:59:18.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/962910/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/10/962910/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/4"}}},"commit":{"parents":[{"commit":"05068ff8f60506874193f3d68ec8645ce9fcc717","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/05068ff8f60506874193f3d68ec8645ce9fcc717"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 09:08:19.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 10:59:08.000000000","tz":120},"subject":"Add support for mistral_api_uwsgi_config in Debian","message":"Add support for mistral_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Mistral API. Therefore, this patch adds a new\nmistral_api_uwsgi_config provider as well as a new\nmistral::wsgi::uwsgi class.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I3e1a7b170c9840b0de83e0eb97c596154790d019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/28118de6f25e6d807f436ffefacd489274e64e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/28118de6f25e6d807f436ffefacd489274e64e24"}]},"branch":"refs/heads/master"},"451856b310115b5dbf1740ec0fbf491ecdce0fff":{"kind":"REWORK","_number":5,"created":"2025-10-03 12:29:25.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/962910/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/10/962910/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/5"}}},"commit":{"parents":[{"commit":"05068ff8f60506874193f3d68ec8645ce9fcc717","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/05068ff8f60506874193f3d68ec8645ce9fcc717"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 09:08:19.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 12:29:14.000000000","tz":120},"subject":"Add support for mistral_api_uwsgi_config in Debian","message":"Add support for mistral_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Mistral API. Therefore, this patch adds a new\nmistral_api_uwsgi_config provider as well as a new\nmistral::wsgi::uwsgi class.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I3e1a7b170c9840b0de83e0eb97c596154790d019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/451856b310115b5dbf1740ec0fbf491ecdce0fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/451856b310115b5dbf1740ec0fbf491ecdce0fff"}]},"branch":"refs/heads/master"},"92ca9b40e299829168a10d15219483b7ec9dbe2a":{"kind":"REWORK","_number":6,"created":"2025-10-03 12:43:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/962910/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/10/962910/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/10/962910/6"}}},"commit":{"parents":[{"commit":"05068ff8f60506874193f3d68ec8645ce9fcc717","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/05068ff8f60506874193f3d68ec8645ce9fcc717"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-03 09:08:19.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-03 12:43:32.000000000","tz":0},"subject":"Add support for mistral_api_uwsgi_config in Debian","message":"Add support for mistral_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Mistral API. Therefore, this patch adds a new\nmistral_api_uwsgi_config provider as well as a new\nmistral::wsgi::uwsgi class.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I3e1a7b170c9840b0de83e0eb97c596154790d019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/92ca9b40e299829168a10d15219483b7ec9dbe2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/92ca9b40e299829168a10d15219483b7ec9dbe2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
