)]}'
{"id":"openstack%2Fpuppet-monasca~654118","triplet_id":"openstack%2Fpuppet-monasca~master~I735f9edb961c500ccb1da8bddd1adcfb3deff41a","project":"openstack/puppet-monasca","branch":"master","topic":"fix-REAME.md","hashtags":[],"change_id":"I735f9edb961c500ccb1da8bddd1adcfb3deff41a","subject":"Fix Repository","status":"MERGED","created":"2019-04-21 05:00:32.000000000","updated":"2019-04-21 05:38:45.000000000","submitted":"2019-04-21 05:38:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654118-1555825125327-332a2ac9","meta_rev_id":"24b1d69e4647d337ac3610f69a9c8740cfb650e0","_number":654118,"virtual_id_number":654118,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-21 05:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-04-21 05:33:45.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-04-21 05:33:45.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-21 05:38:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20287556be75e6a14ecc1204087278ba9326fa45","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 05:00:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7281ac6ea0ea5b57bdf8578f3f1c186cc9be5f1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:04:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/18/654118/1/check/puppet-openstack-lint-ubuntu-bionic/c989841/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ed5eddc0158530af61537881d81b92a556d4f5ac","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 05:33:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48159ee5c64b49bf2c4e33ceb699823980b44382","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:33:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"495e326edefbd1223ad9a959eecedcac7c74563f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:38:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"48f8e27cc462576574c62b61c42342a6280a97b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:38:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/18/654118/1/gate/puppet-openstack-lint-ubuntu-bionic/d977c2a/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b278f17759b68798b6ad3659c8da8751bec99cfc","revisions":{"b278f17759b68798b6ad3659c8da8751bec99cfc":{"kind":"REWORK","_number":1,"created":"2019-04-21 05:00:32.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/18/654118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-monasca","ref":"refs/changes/18/654118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/18/654118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/18/654118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/18/654118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-monasca refs/changes/18/654118/1"}}},"commit":{"parents":[{"commit":"9fcc7c9639b29f074ee2f0ccd46ac20614487c3e","subject":"Replace git.openstack.org with opendev.org/openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/9fcc7c9639b29f074ee2f0ccd46ac20614487c3e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 04:59:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 04:59:09.000000000","tz":480},"subject":"Fix Repository","message":"Fix Repository\n\nChange-Id: I735f9edb961c500ccb1da8bddd1adcfb3deff41a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/b278f17759b68798b6ad3659c8da8751bec99cfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/b278f17759b68798b6ad3659c8da8751bec99cfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
