)]}'
{"id":"openstack%2Fpuppet-monasca~680258","triplet_id":"openstack%2Fpuppet-monasca~master~I8516a261ee7f6ed88d7d2b2bdb611ac8c4bef994","project":"openstack/puppet-monasca","branch":"master","topic":"prep-release-train-4","hashtags":[],"change_id":"I8516a261ee7f6ed88d7d2b2bdb611ac8c4bef994","subject":"Prepare Train M4","status":"MERGED","created":"2019-09-05 01:06:01.000000000","updated":"2019-09-06 14:13:35.000000000","submitted":"2019-09-06 14:13:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680258-1567779215035-f36e0462","meta_rev_id":"14abadf155b5e67f16c77215e08b70a4bd96ddf5","_number":680258,"virtual_id_number":680258,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-09-06 14:13:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-09-06 14:01:51.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-09-06 14:01:51.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 14:01:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-06 14:13:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c945e2585ad949600bf2e988f6ad821bf9a5d3aa","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-05 01:06:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"013f0bec5aa52a422b343c552b651b24a42682b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 01:53:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/30ca26a0d3784a2e9fd19613d71b0167 : SUCCESS in 2m 58s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61394d698b8f4efca0f360b8e5c26e4a : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0eb0ad4d3572165a0a1ebe6737d1d8e7ddf08512","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-06 14:01:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"266c571fbdf8af72b3458c3fabae62962ad157f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:03:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42dc23e2f65da6f9278ff43502331daaafe71f96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:13:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/63d9a5154d7342feb77751b64b20d90f : SUCCESS in 2m 15s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55e09265dd60474c9ec83c52c4ff24cc : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b2a4aaa2aed5e20d743079bfb57e8ecf912de250","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:13:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c76be66134d73b34a602ac212c490f51a5ed6dc","revisions":{"7c76be66134d73b34a602ac212c490f51a5ed6dc":{"kind":"REWORK","_number":1,"created":"2019-09-05 01:06:01.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/58/680258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-monasca","ref":"refs/changes/58/680258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/58/680258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/58/680258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/58/680258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-monasca refs/changes/58/680258/1"}}},"commit":{"parents":[{"commit":"d48e9f889a824bc12a5e20c9c68c1b45d3766aa9","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/d48e9f889a824bc12a5e20c9c68c1b45d3766aa9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-05 01:04:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-05 01:04:49.000000000","tz":480},"subject":"Prepare Train M4","message":"Prepare Train M4\n\nUpdate the version metadata for Train milestone 4\n\nChange-Id: I8516a261ee7f6ed88d7d2b2bdb611ac8c4bef994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/7c76be66134d73b34a602ac212c490f51a5ed6dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/7c76be66134d73b34a602ac212c490f51a5ed6dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
