)]}'
{"id":"openstack%2Fpuppet-monasca~684466","triplet_id":"openstack%2Fpuppet-monasca~master~I345e156fc5a8d5c39fab9d4e9c9a76d1bcedbc98","project":"openstack/puppet-monasca","branch":"master","topic":"prep-release-train-rc1","hashtags":[],"change_id":"I345e156fc5a8d5c39fab9d4e9c9a76d1bcedbc98","subject":"Prepare Train RC1","status":"MERGED","created":"2019-09-25 05:46:52.000000000","updated":"2019-09-27 01:59:16.000000000","submitted":"2019-09-27 01:59:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684466-1569549556079-d890bd4f","meta_rev_id":"ceeca16ba6aef9f8df6edff2e4455040e5dca669","_number":684466,"virtual_id_number":684466,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 01:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:48:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:48:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:48:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 01:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2dd470fb562fe42463300864852e9f7291a0cbf6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-25 05:46:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1ca5e6484842374d4b63250ee49286b5b5d681a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 05:52:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/1243e90f4946467a86184f7f6369c0f4 : SUCCESS in 3m 17s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72b88afdd56a4095b243d3de89c97a62 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"21a74695eb5eb3c2e77b64ee278d934d03ae533c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:48:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab39547813759cce093d22c2a27e3b63ffdf5b6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:48:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1093dd62d1a042e118d5e32b060cf7b7cc84b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:59:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/5c574d6fc6924b5dbfc6818dc29915f9 : SUCCESS in 2m 49s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/401eb5f1e6f94465bcbea696320105bf : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6a0e06e2c3553ef36228aee3198238c672ed364f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:59:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84b2d1443475f46637f8fbf8dc617855d485dc8d","revisions":{"84b2d1443475f46637f8fbf8dc617855d485dc8d":{"kind":"REWORK","_number":1,"created":"2019-09-25 05:46:52.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/66/684466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-monasca","ref":"refs/changes/66/684466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/66/684466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/66/684466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/66/684466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-monasca refs/changes/66/684466/1"}}},"commit":{"parents":[{"commit":"7c76be66134d73b34a602ac212c490f51a5ed6dc","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/7c76be66134d73b34a602ac212c490f51a5ed6dc"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:43.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:43.000000000","tz":480},"subject":"Prepare Train RC1","message":"Prepare Train RC1\n\nUpdate the version metadata for Train milestone RC1\n\nChange-Id: I345e156fc5a8d5c39fab9d4e9c9a76d1bcedbc98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/84b2d1443475f46637f8fbf8dc617855d485dc8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/84b2d1443475f46637f8fbf8dc617855d485dc8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
