)]}'
{"id":"openstack%2Fpuppet-monasca~687038","triplet_id":"openstack%2Fpuppet-monasca~stable%2Ftrain~Ibabf694025fa6e5efef1925941c8e7b0dad88148","project":"openstack/puppet-monasca","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"Ibabf694025fa6e5efef1925941c8e7b0dad88148","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-10-07 13:34:32.000000000","updated":"2019-10-10 15:14:21.000000000","submitted":"2019-10-10 15:14:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687038-1570720461572-0f5a3271","meta_rev_id":"36d63f3e494e007432633de86da69d49dd5d6272","_number":687038,"virtual_id_number":687038,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 15:14:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-10-07 21:06:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-10-10 14:36:56.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 21:06:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-10 14:36:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-10-10 15:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f906eea185721dfec954d43b3b4be1e36f32b1a1","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-07 13:34:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87eb207c9442b2889f97cd4fc8761cb6878ca8b2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-07 21:06:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ebeb04746806afe60c57b5818ccf49d05ce8369","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-09 19:32:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d9ce9e0eecd5a5cd0fc71882c68acbffefa34aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 19:43:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c74b43ccc7c5422b86bc272e3c724a72 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cad92c50576a4f0bb32946436fa87c3a : SUCCESS in 5m 20s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/be1c37d16a04470f8d0a925ef117130a : FAILURE in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"69ab93b2cb534dd967a3325789a9357fdf24f999","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-10 14:36:56.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4397be4b28807fe359af29d3700f390bec3f3d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:59:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c01bdda7ef514e308c6fcdbf7dd7c48f : SUCCESS in 3m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af816026bdbf48b1a3227abbcb74a5cb : SUCCESS in 4m 32s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/80ad1f8d0c0f421cb3e5991237f9296c : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8aa8fc94f07036290fd4b996aeb671433582eaee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:59:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c98ff87b92aba09fe0e285a77bcc8c7bd93dfbb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:14:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e54bdf04a42c10b7e79f75bdc2e97711188cabe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:14:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25e63236df344fc2b40677d50e69cb64 : SUCCESS in 5m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03e04fa91539490fa4ef9ff4624e2006 : SUCCESS in 4m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/281b14b97b7d402abc19e7e302cd107e : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f41d2a652427d609f4b2ad93cca10e982f2012a","revisions":{"9f41d2a652427d609f4b2ad93cca10e982f2012a":{"kind":"REWORK","_number":1,"created":"2019-10-07 13:34:32.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/38/687038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-monasca","ref":"refs/changes/38/687038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/38/687038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/38/687038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/38/687038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-monasca refs/changes/38/687038/1"}}},"commit":{"parents":[{"commit":"9f5bf736a77acbcfe3c53c009429a4ad91d41914","subject":"Merge \"Update the constraints url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/9f5bf736a77acbcfe3c53c009429a4ad91d41914"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-07 13:34:29.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-07 13:34:29.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: Ibabf694025fa6e5efef1925941c8e7b0dad88148\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/9f41d2a652427d609f4b2ad93cca10e982f2012a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/9f41d2a652427d609f4b2ad93cca10e982f2012a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
