)]}'
{"id":"openstack%2Fpuppet-monasca~719550","triplet_id":"openstack%2Fpuppet-monasca~master~Idef1967404d603863425cafb08c51a72a929726e","project":"openstack/puppet-monasca","branch":"master","topic":"prep-release-ussuri-3.1","hashtags":[],"change_id":"Idef1967404d603863425cafb08c51a72a929726e","subject":"Prepare Ussuri M3 again","status":"MERGED","created":"2020-04-13 15:03:43.000000000","updated":"2020-04-13 15:23:49.000000000","submitted":"2020-04-13 15:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719550-1586791429569-6db76d6a","meta_rev_id":"72e0465441332601d53b6c28fb76d58963fd966b","_number":719550,"virtual_id_number":719550,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 15:23:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-13 15:11:50.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-13 15:11:50.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 15:11:50.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-13 15:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4d7478fa2f4fb4e05337284b2550f6b46659da6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-13 15:03:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c74da39869f8672f716fb0f58dd88fc0fbf3993","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-13 15:11:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9001db8a3be8ff3325798032603fcea904870335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:17:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/964177b330c749548d0bfc5a9c03d3a4 : SUCCESS in 3m 55s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b496a4802556452f908c6bf81b5384c5 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9fc5bf7f54bf501be57a202ca9be61651764a202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:18:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a30e058626297ccac898a48ab62031f11e508248","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:23:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8557ec55ef22ae1f2095a0de853b128f20576164","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:23:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/6933e053cfd348f094f88d5a858b88e0 : SUCCESS in 3m 57s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2ce59417a744d1f8872c17a661f926e : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50f644ffda2624b3fc264a3fae1a4a321ea8b184","revisions":{"50f644ffda2624b3fc264a3fae1a4a321ea8b184":{"kind":"REWORK","_number":1,"created":"2020-04-13 15:03:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/719550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-monasca","ref":"refs/changes/50/719550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/50/719550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/50/719550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/50/719550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-monasca refs/changes/50/719550/1"}}},"commit":{"parents":[{"commit":"bee295069c9e6254af6cf62f0ccfca7cc8bed95f","subject":"Resolve duplicated requirement names in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/bee295069c9e6254af6cf62f0ccfca7cc8bed95f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-13 15:03:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-13 15:03:34.000000000","tz":540},"subject":"Prepare Ussuri M3 again","message":"Prepare Ussuri M3 again\n\nUpdate the version metadata to recreate release for Ussuri milestone M3\n\nChange-Id: Idef1967404d603863425cafb08c51a72a929726e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/50f644ffda2624b3fc264a3fae1a4a321ea8b184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/50f644ffda2624b3fc264a3fae1a4a321ea8b184"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
