)]}'
{"id":"openstack%2Fpuppet-monasca~756108","triplet_id":"openstack%2Fpuppet-monasca~master~I9ef9dc1d03d8e35e1381bc055d6c554eea436bc3","project":"openstack/puppet-monasca","branch":"master","topic":"config-lint","hashtags":[],"change_id":"I9ef9dc1d03d8e35e1381bc055d6c554eea436bc3","subject":"Update comment to fix lint","status":"MERGED","created":"2020-10-05 15:22:30.000000000","updated":"2020-10-13 15:54:21.000000000","submitted":"2020-10-13 15:54:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756108-1602604461593-577aa814","meta_rev_id":"122570496e48b0057ae309182bfdef1845278185","_number":756108,"virtual_id_number":756108,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-13 15:54:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-13 15:33:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-12 13:52:18.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-10-10 15:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-10-13 15:33:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-10 15:31:29.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2020-10-12 13:52:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-13 15:33:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-13 15:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1627e0bf243b3dc878266677d332f65223ebdad7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 15:22:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d19e6562d4006dded0e886e12698ea9760397a7f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 15:22:58.000000000","message":"Topic set to config-lint","accounts_in_message":[],"_revision_number":1},{"id":"a0bd41942eb196b5e439e7351b1897796b81edce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 15:54:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/517c128a2d294f87a38d08936480c336 : FAILURE in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ab118b3638c4c5591f450e4d3d660f1 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/abb35038728a4ddfbe18624071abc387 : SUCCESS in 9m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f44929bff2684ec28216935a52501190 : SUCCESS in 6m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31e93e337a094b28afa7c077dcd1f3e4 : SUCCESS in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56036bb8899f0fc5a6a08cda4c31a589e41c885d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-06 12:04:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4e74697f9d3cf5bcd133302b5b7f064402afbd9b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 03:32:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"985225d602b1f61cf222247895a5bf6250f50610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 03:45:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae71f2adbd3d451c91ffa50444ea1c8c : SUCCESS in 3m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c98f59a0b46e43e58de6cfe45eb00169 : SUCCESS in 3m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c49371c07e0846a8b3307759336e67d8 : SUCCESS in 10m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d28f607d11e4883bd4234237333f6c5 : SUCCESS in 7m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3373fc7912e14acdbee2ff02345b34b8 : SUCCESS in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75080ba1f379837f1f98c467adfb30e823b0bc7d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2020-10-10 15:31:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c5d732d13708ea9d820c98af93f41c543fe65b7f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-12 13:52:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6b7285242ac6798dc6524a8bf970acf0ce645857","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-13 15:33:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"17de2eeaa5ad23c409cc7421dac907ba20e148d2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-13 15:33:55.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"674b990af1164d9669ab6b646433c42fd98ed109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 15:34:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30c824ebfdbcb09becb86f9a608d7ce6b7c1e1a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 15:54:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a48ab78833a9413a938fa3dcaadd786b : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18eebc2f2c0840e2b3c2e4037aebaeea : SUCCESS in 3m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d77630f226f5420caeb43f432d29fc40 : SUCCESS in 9m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7c254f566464ef7bd3e56c60a4edbad : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":2},{"id":"769ca1e490d3743c401f93809916d9288390c810","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 15:54:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a6589ed59a0e40069b7e51919db35ab20f80f9a8","revisions":{"bb71fa3e4a6317b78a8d1e76121caf427f485295":{"kind":"REWORK","_number":1,"created":"2020-10-05 15:22:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/756108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-monasca","ref":"refs/changes/08/756108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/08/756108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/08/756108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/08/756108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-monasca refs/changes/08/756108/1"}}},"commit":{"parents":[{"commit":"e69975ced71532194822255e9b3a3aafb3b39bf3","subject":"Use openstack_spec_helper from zuul checkout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/e69975ced71532194822255e9b3a3aafb3b39bf3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 15:21:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 15:21:59.000000000","tz":540},"subject":"Update comment to fix lint","message":"Update comment to fix lint\n\nmanifests/config.pp:7:WARNING: No matching class parameter for\ndocumentation of monasca::config::xxx_config\n\nChange-Id: I9ef9dc1d03d8e35e1381bc055d6c554eea436bc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/bb71fa3e4a6317b78a8d1e76121caf427f485295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/bb71fa3e4a6317b78a8d1e76121caf427f485295"}]},"branch":"refs/heads/master"},"a6589ed59a0e40069b7e51919db35ab20f80f9a8":{"kind":"REWORK","_number":2,"created":"2020-10-10 03:32:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/756108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-monasca","ref":"refs/changes/08/756108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/08/756108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/08/756108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-monasca refs/changes/08/756108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-monasca refs/changes/08/756108/2"}}},"commit":{"parents":[{"commit":"a726f3e07bd659bb2c24611434a4a6d9d17393fc","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/a726f3e07bd659bb2c24611434a4a6d9d17393fc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 15:21:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-10 03:32:08.000000000","tz":540},"subject":"Update comment to fix lint","message":"Update comment to fix lint\n\nmanifests/config.pp:7:WARNING: No matching class parameter for\ndocumentation of monasca::config::xxx_config\nmanifests/keystone/auth.pp:68:WARNING: No matching class parameter for\ndocumentation of monasca::keystone::auth::region\nmanifests/keystone/auth.pp:65:WARNING: No matching class parameter for\ndocumentation of monasca::keystone::auth::port\nmanifests/keystone/auth.pp:25:WARNING: No matching class parameter for\ndocumentation of monasca::keystone::auth::agent_password\nmanifests/keystone/auth.pp:22:WARNING: No matching class parameter for\ndocumentation of monasca::keystone::auth::user_password\nmanifests/keystone/auth.pp:19:WARNING: No matching class parameter for\ndocumentation of monasca::keystone::auth::admin_password\nmanifests/keystone/auth.pp:16:WARNING: No matching class parameter for\ndocumentation of monasca::keystone::auth::agent_name\nmanifests/keystone/auth.pp:12:WARNING: No matching class parameter for\ndocumentation of monasca::keystone::auth::user_name\nmanifests/keystone/auth.pp:9:WARNING: No matching class parameter for\ndocumentation of monasca::keystone::auth::admin_name\nmanifests/persister/config.pp:34:WARNING: No matching defined type\nparameter for documentation of monasca::persister::config::pers_db_user\n\nChange-Id: I9ef9dc1d03d8e35e1381bc055d6c554eea436bc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/a6589ed59a0e40069b7e51919db35ab20f80f9a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-monasca/commit/a6589ed59a0e40069b7e51919db35ab20f80f9a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
