)]}'
{"id":"openstack%2Fpuppet-murano~654121","triplet_id":"openstack%2Fpuppet-murano~master~Id5415e2f4799d1601553d7fcbabd6956085eb2a2","project":"openstack/puppet-murano","branch":"master","topic":"fix-REAME.md","hashtags":[],"change_id":"Id5415e2f4799d1601553d7fcbabd6956085eb2a2","subject":"Fix Repository","status":"MERGED","created":"2019-04-21 05:00:36.000000000","updated":"2019-04-21 05:45:14.000000000","submitted":"2019-04-21 05:45:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654121-1555825514692-332a2ac9","meta_rev_id":"58af6f947be8ff399eaae25f5d6cbe08ac53dce1","_number":654121,"virtual_id_number":654121,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-21 05:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-04-21 05:33:42.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-04-21 05:33:42.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-21 05:45:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb18eff1a8bf45b2fb660f7b21f8207b0fdaa145","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 05:00:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9abbfea0bb8a66e09535cb0122e80b1bc76d5ab8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:04:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/21/654121/1/check/puppet-openstack-lint-ubuntu-bionic/ea75fd1/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1024896d2ec2c49a1c1781615839678bffa397b9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 05:33:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"978ab829b19003313adf6c442f45ec25401b56ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:33:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d147ac10ab078d2af99d8100ce8fb35b7c0b84e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:45:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dfcd5456283caa68ee5be5d9a76e1efb1e7358e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 05:45:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/21/654121/1/gate/puppet-openstack-lint-ubuntu-bionic/54e0575/ : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9b456a19f2dc1d22da751d05e85dc546a3b8547","revisions":{"f9b456a19f2dc1d22da751d05e85dc546a3b8547":{"kind":"REWORK","_number":1,"created":"2019-04-21 05:00:36.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/21/654121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/21/654121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/21/654121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/21/654121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/21/654121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/21/654121/1"}}},"commit":{"parents":[{"commit":"fb4f0e8321a4add5a11702063d1e7ee750dfcb55","subject":"Replace git.openstack.org with opendev.org/openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/fb4f0e8321a4add5a11702063d1e7ee750dfcb55"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 04:59:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 04:59:09.000000000","tz":480},"subject":"Fix Repository","message":"Fix Repository\n\nChange-Id: Id5415e2f4799d1601553d7fcbabd6956085eb2a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/f9b456a19f2dc1d22da751d05e85dc546a3b8547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/f9b456a19f2dc1d22da751d05e85dc546a3b8547"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
