)]}'
{"id":"openstack%2Fpuppet-murano~674984","triplet_id":"openstack%2Fpuppet-murano~master~I6b1a478f0fe628f00a0ecaf20c4c19f5d63c689e","project":"openstack/puppet-murano","branch":"master","topic":"fix-sphinx","hashtags":[],"change_id":"I6b1a478f0fe628f00a0ecaf20c4c19f5d63c689e","subject":"Sync Sphinx requirement","status":"MERGED","created":"2019-08-07 01:13:45.000000000","updated":"2019-08-07 02:29:41.000000000","submitted":"2019-08-07 02:29:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674984-1565144981879-cd85f96c","meta_rev_id":"66ac1c2db7a3ca0a52ea8769b1c016b7f999cc94","_number":674984,"virtual_id_number":674984,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-07 02:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-08-07 02:25:57.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-08-07 02:25:57.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 02:29:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87629860683c1674d5a90f2c48a27239956d28f9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-07 01:13:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"771ffc5433418afb55a532a8f4082b492bc967c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 01:17:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://logs.opendev.org/84/674984/1/check/build-openstack-releasenotes/7b4c3cc/html/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"77902f612d069a875aeb9ce659f0e09501fdcb36","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-07 02:25:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff8e14c901a972560927fbbdb7b90273e0aa5310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:26:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e256a7cc31b2f6ba056be49fabe055b98c924b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:29:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2217caefe1aa17e84107e8861c6c2f58ed59d640","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:29:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://logs.opendev.org/84/674984/1/gate/build-openstack-releasenotes/fd94437/html/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a132c26abb4a018763fc945177fe4f59cf943155","revisions":{"a132c26abb4a018763fc945177fe4f59cf943155":{"kind":"REWORK","_number":1,"created":"2019-08-07 01:13:45.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/84/674984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/84/674984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/84/674984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/84/674984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/84/674984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/84/674984/1"}}},"commit":{"parents":[{"commit":"87117d2d9571a3f8cb1f77057e5a6a883984bc56","subject":"Prepare Train M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/87117d2d9571a3f8cb1f77057e5a6a883984bc56"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-07 01:10:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-07 01:10:09.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\nSync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n\nChange-Id: I6b1a478f0fe628f00a0ecaf20c4c19f5d63c689e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/a132c26abb4a018763fc945177fe4f59cf943155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/a132c26abb4a018763fc945177fe4f59cf943155"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
