)]}'
{"id":"openstack%2Fpuppet-murano~684498","triplet_id":"openstack%2Fpuppet-murano~master~I755d888f5c083d839793e880ba294d38dd40e7e4","project":"openstack/puppet-murano","branch":"master","topic":"prep-release-train-rc1","hashtags":[],"change_id":"I755d888f5c083d839793e880ba294d38dd40e7e4","subject":"Prepare Train RC1","status":"MERGED","created":"2019-09-25 05:46:58.000000000","updated":"2019-09-27 01:58:57.000000000","submitted":"2019-09-27 01:58:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684498-1569549537821-e03b07bf","meta_rev_id":"ce28b2d9c944bcd878fc0096dcc17a8b05ff3fe8","_number":684498,"virtual_id_number":684498,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 01:58:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:48:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:48:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:48:40.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 01:58:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c0df781df04690cd2621a9f1193755fa49e04a1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-25 05:46:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e499bd359d5462cf34393cc68f8bb2e6d566700f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 05:56:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/1162efdaeca049b4ae42ddfd5d571049 : SUCCESS in 2m 08s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/414929d8333d42b29ed0a3ee6dbd8bd8 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cb3a95e2bba2680e5e7ee5d9bd6ffa67a4f5229e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:48:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b67153c251b10af80e6fa04456a0f6a13cc628fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:48:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cfe21c99188cef91b3c218bf3ca7da7d27339fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:58:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"324b85bd65955364ba28ab99697ad920ba7b9a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:58:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/3dda982270424a21ab14f722906f484c : SUCCESS in 3m 37s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72c1b7f9b05b4145818a351358be4d33 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ec08ff06b6bae859e27436981c7cb6eb65c4446","revisions":{"2ec08ff06b6bae859e27436981c7cb6eb65c4446":{"kind":"REWORK","_number":1,"created":"2019-09-25 05:46:58.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/98/684498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/98/684498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/98/684498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/98/684498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/98/684498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/98/684498/1"}}},"commit":{"parents":[{"commit":"699976d96e5032b468bba3bc6047dd905bd81e8e","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/699976d96e5032b468bba3bc6047dd905bd81e8e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:43.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:43.000000000","tz":480},"subject":"Prepare Train RC1","message":"Prepare Train RC1\n\nUpdate the version metadata for Train milestone RC1\n\nChange-Id: I755d888f5c083d839793e880ba294d38dd40e7e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/2ec08ff06b6bae859e27436981c7cb6eb65c4446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/2ec08ff06b6bae859e27436981c7cb6eb65c4446"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
