)]}'
{"id":"openstack%2Fpuppet-murano~725175","triplet_id":"openstack%2Fpuppet-murano~master~I536d77e157d96211e317297cefd321d6a7d6a98f","project":"openstack/puppet-murano","branch":"master","topic":"autorequire","hashtags":[],"change_id":"I536d77e157d96211e317297cefd321d6a7d6a98f","subject":"Use anchor to require necessary packages","status":"MERGED","created":"2020-05-04 06:11:13.000000000","updated":"2020-05-04 13:22:33.000000000","submitted":"2020-05-04 13:22:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725175-1588598553355-3c423189","meta_rev_id":"1148b708202d95c1e1ce148fc81c690529e9fffa","_number":725175,"virtual_id_number":725175,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 13:22:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-04 13:02:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-04 07:02:31.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-04 13:02:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 07:02:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-04 13:02:47.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-04 13:22:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f378deddf5e3fa41aaf51928de860f04befec071","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-04 06:11:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7906a29535470611821e2d6c131185908e1a460f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-04 06:12:06.000000000","message":"Topic set to autorequire","accounts_in_message":[],"_revision_number":1},{"id":"58777392d725c8e6e40055eebc5516c3d6c80624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 06:43:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2cb2a625ec0412baf554c18d9cc2e2e : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c15b3df35e24f52bcfa713a6f208afe : SUCCESS in 7m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/681a65c74a1749a6ae716cf91ec7815a : SUCCESS in 5m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/49bffd49a2944b6aa43a3191f0a2b4c0 : SUCCESS in 14m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4d546190887240b7b66d5d79e1b59936 : SUCCESS in 12m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/524f7eb5197b456f8c3f38b1e03bbf51 : SUCCESS in 9m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c6a0e4b804c4560933389685f9d4e63 : SUCCESS in 9m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/66153579145b4557b9a47d320069d979 : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":1},{"id":"148476a19cb445591a1c8fa60e51a9b6bec04284","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-04 07:02:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8006045e4eccac8664e09ebf7acb8147ac0f87b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-04 13:02:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5df9954dca54461d158dd8bd30b0c450a17c0cb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:02:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fe2310dba37eb9a7167c3145d103fb4aa56032b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:22:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1148b708202d95c1e1ce148fc81c690529e9fffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:22:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b2311a7607244d0b6347d035d56303f : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46ac5b904e4f465f8e8170aceddea5e1 : SUCCESS in 7m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c2d13cb726c4c2680e1299c28da8433 : SUCCESS in 5m 53s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/817602762076407dbbbb1c0a5e6f39ec : SUCCESS in 14m 49s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/75c77a11b7cf4dea9ace519c3564e480 : SUCCESS in 13m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4915e77a1e7049f0b6cf5bd2e0b27a8f : SUCCESS in 10m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b8c44bcd83c04cd99cbddb866e641df6 : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a503ac22c307c6732471598a6c666519854b41fa","revisions":{"a503ac22c307c6732471598a6c666519854b41fa":{"kind":"REWORK","_number":1,"created":"2020-05-04 06:11:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/725175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/75/725175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/75/725175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/75/725175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/75/725175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/75/725175/1"}}},"commit":{"parents":[{"commit":"aab4bb05015051164c2acb6810c1a9472cf8815a","subject":"Runs on Debian 10.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/aab4bb05015051164c2acb6810c1a9472cf8815a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-04 06:11:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-04 06:11:01.000000000","tz":540},"subject":"Use anchor to require necessary packages","message":"Use anchor to require necessary packages\n\n... so that correct packages are required without re-defining them in\nresource implementations.\n\nChange-Id: I536d77e157d96211e317297cefd321d6a7d6a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/a503ac22c307c6732471598a6c666519854b41fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/a503ac22c307c6732471598a6c666519854b41fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
