)]}'
{"id":"openstack%2Fpuppet-murano~749123","triplet_id":"openstack%2Fpuppet-murano~master~I0171e5a73d8d36257656d836a9231249d0b88b84","project":"openstack/puppet-murano","branch":"master","topic":"749123","hashtags":[],"change_id":"I0171e5a73d8d36257656d836a9231249d0b88b84","subject":"Add Puppet Litmus","status":"MERGED","created":"2020-08-31 19:41:20.000000000","updated":"2020-09-11 20:05:21.000000000","submitted":"2020-09-11 20:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749123-1599854721051-8dffc5fa","meta_rev_id":"9b8032fa06995919a9953a391b292f1b95efdf25","_number":749123,"virtual_id_number":749123,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 20:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-11 12:37:10.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-11 19:39:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-11 19:39:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 12:37:10.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-11 19:39:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-11 20:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50c0615ca0605651b97456222cc15c08be627c57","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 19:41:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"888a638e31f48c82638ba4e67855d93f7f6fcd30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 19:41:44.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/puppet-murano on branch master.  The error was:\n\n  Job puppet-openstack-litmus-centos-7 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - puppet-openstack-check-jobs\n        - puppet-openstack-module-unit-jobs\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - puppet-openstack-litmus-centos-7\n          - puppet-openstack-litmus-xenial:\n              voting: false\n      gate:\n        jobs:\n          - puppet-openstack-litmus-centos-7\n\n  in \"openstack/puppet-murano/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"ca6c956075c51fc8216a351498b57efb676d90fd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 20:07:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"281d252c2c7466551c3900b0101c1c436a36ae22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 20:31:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/860578fb38f7439a930798cc8889026f : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc93f0d7946e44b0a3cc0ae5270cc3c7 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/86f31d995bcd42e7992a3c6b01cc20ba : SUCCESS in 10m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65f3f9c0bb88455d94f05242978582d3 : SUCCESS in 9m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b845a48e54e0403b9b4ac5b71d4086e5 : SUCCESS in 10m 54s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/fb3a841f08e5422fb98d2603b28679df : SUCCESS in 16m 35s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ac2bd6e8adc4bec9b94bfd1e6c69699 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"ec91170393a4dfd56c3b00d11f2b60f8c1bd0289","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:20:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9075b9d5830c0e1044509308d040a9c4b9f75fde","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 18:28:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"92c70824ba225b1a16e2586e11decb140d7cd431","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-09 21:27:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99815c12ec000dc2d0d24272b8cb83c7cc336b56","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-09 22:56:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ca2452c40e75c02e908bde2af6341fccf5da5c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:46:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/971b5ebcf1244fa9b49e6f1daea7ce5f : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7357f640045436fa587b5d2405a7284 : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0ac6cfcca82f44a4ac0dc4db9f4cfd76 : SUCCESS in 10m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d47f19e95b7443190ff2ca12711e4d7 : SUCCESS in 8m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d1d04d918864a0697775b9e8757cadd : SUCCESS in 9m 26s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/40bddcc3245e4039812ad527e1e0338f : FAILURE in 16m 54s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d17a2e41c30447588fe5733e3caf2c61 : FAILURE in 14m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9e6901e9eef53210463c5f428bb7d8767a6578fe","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-10 22:21:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ac4c4703edfbde204c0b384552a3124024ff3ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 22:46:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aaf73a18bf544e280e5108d775b5f99 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/026c9d65115a4b0da7f8a48c5b559b53 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/724b2c8c203441cbb6fdeb755e4e030c : SUCCESS in 10m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b5ef62c5301467bba39db191bb243bb : SUCCESS in 8m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1967b22fa8654a84be3d966f7123e36c : SUCCESS in 9m 10s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d9247338cd33491b9206fecd2ca2fd6e : FAILURE in 17m 42s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a40a9f9198e1454ab683b676584a370e : FAILURE in 12m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b472acdd9df809098070e0e24b2759111126d991","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-11 09:45:00.000000000","message":"Patch Set 2:\n\n2020-09-10 22:44:51.068779 | centos-8 |        \u001b[1;31mError: /Stage[main]/Apache::Mod::Wsgi/Apache::Mod[wsgi]/Package[mod_wsgi]/ensure: change from \u0027purged\u0027 to \u0027present\u0027 failed: Execution of \u0027/bin/dnf -d 0 -e 1 -y install mod_wsgi\u0027 returned 1: Error: Unable to find a match: mod_wsgi\u001b[0m","accounts_in_message":[],"_revision_number":2},{"id":"7f1c288b9e8e46c9bf5589f1ad218346dc9ed046","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-11 09:45:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9f681b93c3a56c21b02d6fa5ed5dd0687b875404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 10:07:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/276fadcfab6f4b28895a3b0f2d3e3321 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5464e8696d754769a6b77b3051366444 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/723b70c5569b48699d155475ed1ff82e : SUCCESS in 10m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6cb2cc3c3124ae98651903b5c16342b : SUCCESS in 9m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf614ed466d543a2a1d0ae826bfaab51 : SUCCESS in 10m 14s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/337dcb4732534ff5befaeac77f8269fd : SUCCESS in 18m 07s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/512a55a6ce954f06a5da13d4d038717e : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2faa26b8be27558c4147102d4ac21915541072b3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-11 10:14:10.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"35239b8c3c87a653a9da52df8710a591df76ff0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 10:43:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe0d1c04a90f4d0bad5313ec96698ae2 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a368804d0a84948ba9690d3a2892e2a : SUCCESS in 4m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cdf1bc0aa2304f388d7a6a66dbb525eb : SUCCESS in 10m 26s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/251903a6dab84f048474cd1e21e2a62d : SUCCESS in 15m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59c992fe99c245bd8f6641f727b760b4 : SUCCESS in 13m 49s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/3009c0ad39a84a9f8c0d62d89430a3f5 : SUCCESS in 19m 21s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0948d595cc9466c82d7152ff25ac670 : FAILURE in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"caef9afdca7550d8d4cb58bc03519348ce088761","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-11 12:37:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"45d73b9cf882a1cdcfc8c7fb87d4162759ae49b1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-11 19:39:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bfba4f31b20fe4f5084c38d0521009a834911fe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 19:39:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cb6bfba82c7fb519d84ba78cb9fb4cd5f64de2f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:05:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/070c74f879944aa39f3e23d3c6daf652 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/967a4918a6264e0c8a91ecf9374c8f95 : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/16e7b90102f34b92b6d15c9eee2f3cb5 : SUCCESS in 11m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35c75a5256814d51bf9fcf4e50cc63fe : SUCCESS in 10m 57s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f5a26ee96b2e4a5db901d8c67b3a496c : SUCCESS in 21m 00s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c7194bcdfc9412eb4fd0aacfcf87fe8 : FAILURE in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"389ff9b09f5554a2cc37826d0d4b350b41a68983","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:05:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"97b56d27ee266a5f8c4eb0dfe0624f8b8143fd50","revisions":{"d0ee929f9f887905d452ca87154524ab63068f10":{"kind":"REWORK","_number":1,"created":"2020-08-31 19:41:20.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/23/749123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/23/749123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/1"}}},"commit":{"parents":[{"commit":"dea6dcd9616ff6fa8ebfb2b0e4c709aa4bf20ef5","subject":"Merge \"Use openstack_spec_helper from zuul checkout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/dea6dcd9616ff6fa8ebfb2b0e4c709aa4bf20ef5"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:41:12.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:41:12.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: I0171e5a73d8d36257656d836a9231249d0b88b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/d0ee929f9f887905d452ca87154524ab63068f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/d0ee929f9f887905d452ca87154524ab63068f10"}]},"branch":"refs/heads/master"},"7714ab30582a191f640700270a4b0fca59c64b36":{"kind":"REWORK","_number":2,"created":"2020-08-31 20:07:19.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/23/749123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/23/749123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/2"}}},"commit":{"parents":[{"commit":"dea6dcd9616ff6fa8ebfb2b0e4c709aa4bf20ef5","subject":"Merge \"Use openstack_spec_helper from zuul checkout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/dea6dcd9616ff6fa8ebfb2b0e4c709aa4bf20ef5"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:41:12.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 20:07:13.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: I0171e5a73d8d36257656d836a9231249d0b88b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/7714ab30582a191f640700270a4b0fca59c64b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/7714ab30582a191f640700270a4b0fca59c64b36"}]},"branch":"refs/heads/master"},"a0b25ca038ce350fd52126f3802e3e3bdb75950c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-11 09:45:59.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/23/749123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/23/749123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/3"}}},"commit":{"parents":[{"commit":"f47617c2d79c433c9a7c1bd7277187aa340db17b","subject":"Add openstack_integration::apache to integration manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/f47617c2d79c433c9a7c1bd7277187aa340db17b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:41:12.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-09-11 09:45:59.000000000","tz":0},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: I0171e5a73d8d36257656d836a9231249d0b88b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/a0b25ca038ce350fd52126f3802e3e3bdb75950c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/a0b25ca038ce350fd52126f3802e3e3bdb75950c"}]},"branch":"refs/heads/master"},"97b56d27ee266a5f8c4eb0dfe0624f8b8143fd50":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-09-11 10:14:10.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/23/749123/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/23/749123/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/23/749123/4"}}},"commit":{"parents":[{"commit":"f47617c2d79c433c9a7c1bd7277187aa340db17b","subject":"Add openstack_integration::apache to integration manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/f47617c2d79c433c9a7c1bd7277187aa340db17b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:41:12.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-09-11 10:14:09.000000000","tz":0},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/751233\nChange-Id: I0171e5a73d8d36257656d836a9231249d0b88b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/97b56d27ee266a5f8c4eb0dfe0624f8b8143fd50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/97b56d27ee266a5f8c4eb0dfe0624f8b8143fd50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
