)]}'
{"id":"openstack%2Fpuppet-murano~756788","triplet_id":"openstack%2Fpuppet-murano~stable%2Fvictoria~I451842c77e2b51d77fb86c0270bec6100f8a53f5","project":"openstack/puppet-murano","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I451842c77e2b51d77fb86c0270bec6100f8a53f5","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","status":"MERGED","created":"2020-10-08 14:42:41.000000000","updated":"2020-10-12 17:22:16.000000000","submitted":"2020-10-12 17:20:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756788-1602523235070-473da3c1","meta_rev_id":"ca84ebe756e325ef220745b6ad2bbf214aec3712","_number":756788,"virtual_id_number":756788,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 17:20:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-08 14:48:30.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-10-12 15:41:40.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-10-12 15:41:40.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 14:48:30.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-12 15:41:40.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-12 17:20:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6019bba286cb1fbffa1c1d4d4229506ccef82b59","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-10-08 14:42:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"389ff7eb5c9911916a05dca1b11631f7373e8d96","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-08 14:48:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dfa2be9ebe1e77b48aa54bcf6d1ce5c3ef729f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 15:40:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df1df687a9c448e0a2ca891c7d8cf423 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/440ee7d5b8de4bc7a7f8bdd002f22d41 : SUCCESS in 4m 39s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/25091259fd6846e889792c40ef58d30e : SUCCESS in 10m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a21f29ac786741d696813990415da444 : SUCCESS in 8m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0c77e5eca694a20bd31be376b1464cc : SUCCESS in 9m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/894cfb5585b649d6aa1515d6b66e1653 : SUCCESS in 5m 00s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/687ac6fe62974dcfaa28823b77c056e1 : FAILURE in 8m 16s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/defbc38dc0c246d2ae3f49999fae8bf5 : FAILURE in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"416bd48b38c35ded881ed3cdbf62c8e9aa798c62","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-12 15:41:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb65239369667dbb249d78b015d02041fd3f3dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 16:54:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d92d637c4bb44a882e648c037cb14e0 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af7cd630d7b944ec909e82fabe01f014 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9efb7fba9458474fb21cf105ebefe8be : SUCCESS in 10m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ef531967a0f4e7aad399522b2bd6c1c : SUCCESS in 8m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f66f96bcdb64fb38b44375ea1d7209f : SUCCESS in 8m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f503c7a0d7874accad29108328935df3 : SUCCESS in 4m 16s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6fc2462130fc451688a17bed3c1a0e78 : SUCCESS in 21m 23s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a700936d0e842e899a2d01713673592 : FAILURE in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"163d2d47f14b8ed8df7b47bd401a7c12c0c0406e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 16:59:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e077ac081c7356835c31bd0e1ec403ab38ac0692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 17:20:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d95ef9bc65f948b188ea20e12f73b954 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b611056009947118e5cc413dfa5d075 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/eee945e13c6a425d8d2038de1c64e066 : SUCCESS in 17m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41e8951c35f1439f8534772daf21b704 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39a11a129c8f4aacbbf6e473d21a5103 : SUCCESS in 5m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4ce80beb2185486698be0285edc8bbd6 : SUCCESS in 16m 43s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd9288569dc04e4faf799cbb2caa0cdc : FAILURE in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dff8fada71c749222354892a49b49cf501d1ca97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 17:20:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"19c802c61f91c820c6e161b86f3543c2d35bb13a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 17:22:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7cd9203b1834dba8199032cee9092d4 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5f92eaca3bb2387059f58b620bac5af9d6ca6a3","revisions":{"f5f92eaca3bb2387059f58b620bac5af9d6ca6a3":{"kind":"REWORK","_number":1,"created":"2020-10-08 14:42:41.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/88/756788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/88/756788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/88/756788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/88/756788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/88/756788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/88/756788/1"}}},"commit":{"parents":[{"commit":"ccb7cc5be7c6253c3aad824cc04f31d545633a0b","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/ccb7cc5be7c6253c3aad824cc04f31d545633a0b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:42:39.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:42:39.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","message":"Update TOX_CONSTRAINTS_FILE for stable/victoria\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/victoria branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I451842c77e2b51d77fb86c0270bec6100f8a53f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/f5f92eaca3bb2387059f58b620bac5af9d6ca6a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/f5f92eaca3bb2387059f58b620bac5af9d6ca6a3"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
