)]}'
{"id":"openstack%2Fpuppet-murano~756948","triplet_id":"openstack%2Fpuppet-murano~master~I4fbf5bce75940a1816bc2824cf8596b69ccc4f59","project":"openstack/puppet-murano","branch":"master","topic":"missing-deps","hashtags":[],"change_id":"I4fbf5bce75940a1816bc2824cf8596b69ccc4f59","subject":"Include deps class in unit test for mysql","status":"MERGED","created":"2020-10-09 02:00:41.000000000","updated":"2020-10-09 03:08:05.000000000","submitted":"2020-10-09 03:08:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756948-1602212885987-30d2bd40","meta_rev_id":"6846c696b41b69be636f5ba7352dc7d45bf75863","_number":756948,"virtual_id_number":756948,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 03:08:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-09 02:46:59.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-09 02:46:59.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 03:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2494030c6be66531c368f39928627f85c17a604","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75ba4b68004e3dab6a38218b6d1b011139d0d509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:18:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6525d20ce5d7430b9a7bdf178dc92c0d : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5958c373a60940c294388638c33146bd : SUCCESS in 3m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/36cf07d34a19473f89cbb0ae296b3a41 : SUCCESS in 10m 26s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fe719d8d95f4707b3e947e182c097b2 : SUCCESS in 7m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6815ac3a6b243979ff22829a088a6e5 : SUCCESS in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fff796aa2c00a3221e22d2d4eaa6fc62118da10e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:46:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fa2bf7e1eb508da371c970494de73aef0974710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:47:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48a87239683af49d15363adf06a819374084a573","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:08:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"89165024e061767296c57156c993114498d77ae0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:08:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fb1fd05968c4075a09e3a248cbeb970 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0646fa78da4a4cf8ae0cc7769b6f072d : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0ddaa0194d70492d92119a54df8d0846 : SUCCESS in 10m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/379856b7422c4bfd936f13c0e2f6843c : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f912c53ea8c3ad647c36243478f4afe91f95641","revisions":{"8f912c53ea8c3ad647c36243478f4afe91f95641":{"kind":"REWORK","_number":1,"created":"2020-10-09 02:00:41.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/48/756948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/48/756948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/48/756948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/48/756948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/48/756948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/48/756948/1"}}},"commit":{"parents":[{"commit":"1d7a79e9f2e79756c0517d3e6108fd5ce72c0108","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/1d7a79e9f2e79756c0517d3e6108fd5ce72c0108"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"subject":"Include deps class in unit test for mysql","message":"Include deps class in unit test for mysql\n\nChange-Id: I4fbf5bce75940a1816bc2824cf8596b69ccc4f59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/8f912c53ea8c3ad647c36243478f4afe91f95641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/8f912c53ea8c3ad647c36243478f4afe91f95641"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
