)]}'
{"id":"openstack%2Fpuppet-murano~758286","triplet_id":"openstack%2Fpuppet-murano~master~I0c462bd26eacb2555e44a8b045b0466a0d3bb1d5","project":"openstack/puppet-murano","branch":"master","topic":"prep-release-wallaby-start","hashtags":[],"change_id":"I0c462bd26eacb2555e44a8b045b0466a0d3bb1d5","subject":"Bump version for the start of Wallaby","status":"MERGED","created":"2020-10-15 01:43:42.000000000","updated":"2020-10-15 14:22:51.000000000","submitted":"2020-10-15 14:22:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758286-1602771771662-f8940f66","meta_rev_id":"616bb6b670adf3f520e397ebaf7a0806aa82bcc9","_number":758286,"virtual_id_number":758286,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 14:22:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-15 14:06:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-10-15 07:34:09.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-10-15 14:06:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 07:34:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-15 14:06:33.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-15 14:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a06e95ad2474b6285d909b931cdbc59cc648155","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-15 01:43:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc295a729e1be4ab80f1503f8b54815d52fa4017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 01:51:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/236dab20438d47688f3e574b0ae906c8 : SUCCESS in 2m 41s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf3ef71323204941a4aec91e78e4dd3e : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e6a737aa394f388563fb2638ad6beb0bfb637c8e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-15 07:34:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a8cb044eed08a5ee801f54b21921108b841dca3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-15 14:06:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20be63e165138145e551f0f803cefe0a7f6a5cf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:06:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d50005e6fdd077f84f7c46b242863ec75182bee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:22:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec3daab78d561d6e6da52f7bcffc44436829889c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:22:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/4f804d37e04e4f2a97a26244e30a23dd : SUCCESS in 3m 05s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e17a178ff6d14d519f8b384947f17e57 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a78d2ee5f4a58371c7be954c304c873311b09a6","revisions":{"7a78d2ee5f4a58371c7be954c304c873311b09a6":{"kind":"REWORK","_number":1,"created":"2020-10-15 01:43:42.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/86/758286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/86/758286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/86/758286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/86/758286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/86/758286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/86/758286/1"}}},"commit":{"parents":[{"commit":"d6d0b3d4d9c608d9535fb0c3b6246ec4adbbc59c","subject":"Include deps class in unit test for sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/d6d0b3d4d9c608d9535fb0c3b6246ec4adbbc59c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-15 01:42:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-15 01:42:49.000000000","tz":480},"subject":"Bump version for the start of Wallaby","message":"Bump version for the start of Wallaby\n\nBump the major version for the Wallaby release after Victoria has\nbeen branched.\n\nChange-Id: I0c462bd26eacb2555e44a8b045b0466a0d3bb1d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/7a78d2ee5f4a58371c7be954c304c873311b09a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/7a78d2ee5f4a58371c7be954c304c873311b09a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
