)]}'
{"id":"openstack%2Fpuppet-murano~760017","triplet_id":"openstack%2Fpuppet-murano~master~I8d908b3b8fa8fbae67e2dbcebba3791d4837c6fe","project":"openstack/puppet-murano","branch":"master","topic":"fix-source-metadata","hashtags":[],"change_id":"I8d908b3b8fa8fbae67e2dbcebba3791d4837c6fe","subject":"Fix outdated files","status":"MERGED","created":"2020-10-28 03:03:25.000000000","updated":"2020-10-28 03:37:57.000000000","submitted":"2020-10-28 03:37:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760017-1603856277014-beb8d14c","meta_rev_id":"2078532ba3460eae171bf70fd23b7a11ba5e3f14","_number":760017,"virtual_id_number":760017,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-28 03:37:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-28 03:31:35.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-28 03:31:35.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 03:37:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4650a7f135e0f6938febb23b7fcb8a0ee177114","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-28 03:03:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8cf2c0fcac5f77e7a41f34c5b6d011c675ec3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 03:25:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/01d16cf715a94b9db0290662646752bc : SUCCESS in 2m 14s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ead174457504de6a752263224663486 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"92fb2fb87ee35f8fc587602f567ea430ca555084","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-28 03:31:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12663f6f27389cf4729d6e420b6e28d90370403a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 03:31:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64a6ebe1fb636ce32316167c42ba3beed7efeed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 03:37:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f5b762b4fbbd42819f4b6dffd07f3788 : SUCCESS in 1m 57s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b133119eee240c6bedffa8b7fc9e912 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e2fdff9bc129e59205cd176dd82bf03e77e2d3e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 03:37:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b8af48123bf30d3eb250e328136606a632775a1","revisions":{"3b8af48123bf30d3eb250e328136606a632775a1":{"kind":"REWORK","_number":1,"created":"2020-10-28 03:03:25.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/17/760017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/17/760017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/17/760017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/17/760017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/17/760017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/17/760017/1"}}},"commit":{"parents":[{"commit":"7a78d2ee5f4a58371c7be954c304c873311b09a6","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/7a78d2ee5f4a58371c7be954c304c873311b09a6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-28 02:40:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-28 02:40:09.000000000","tz":480},"subject":"Fix outdated files","message":"Fix outdated files\n\nChange-Id: I8d908b3b8fa8fbae67e2dbcebba3791d4837c6fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/3b8af48123bf30d3eb250e328136606a632775a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/3b8af48123bf30d3eb250e328136606a632775a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
