)]}'
{"id":"openstack%2Fpuppet-murano~781979","triplet_id":"openstack%2Fpuppet-murano~master~Ia54c443e57539e8787b471fb627d1b1bd2ac0989","project":"openstack/puppet-murano","branch":"master","topic":"oslo_policy-enforce_scope","hashtags":[],"change_id":"Ia54c443e57539e8787b471fb627d1b1bd2ac0989","subject":"Add support for the oslo_policy/enforce_scope parameter","status":"MERGED","created":"2021-03-21 02:00:04.000000000","updated":"2021-03-22 17:18:21.000000000","submitted":"2021-03-22 17:18:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781979-oslo_policy-enforce_scope","meta_rev_id":"f5fe8c3ecdbccef48f430e8fea9ae80ce0199bf7","_number":781979,"virtual_id_number":781979,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 17:18:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-22 08:19:12.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-22 16:50:42.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2021-03-22 08:20:55.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-22 16:50:42.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-21 02:28:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 08:19:12.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-22 16:50:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"392e7bfd6dad839a3b8b3527b803321900b6673f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-21 02:00:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7057e580c433ec80f2f58c0da50c36bfb56e1cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 02:28:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8514e791b7784683aa0528373fce65fd : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da3f9cf96e6c406884829a4317c56f96 : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/30e1fca92f954337ac7f113644e98129 : SUCCESS in 13m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff4ee903d1564ed5a74de684341a06c8 : SUCCESS in 7m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25bdd4685a6248c18f463b30c490b812 : SUCCESS in 7m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a5df7b324204b608e3e857d502ddc45 : SUCCESS in 3m 28s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/3eec7e47332f42ad9a9840d4c55bcfd4 : SUCCESS in 19m 31s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd4411cf69e846a2a98940c9eedd8371 : FAILURE in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6951380e541db607b9cda3cf2e766ecb475d44a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-22 08:19:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68ba3892a407fa9cbc675c916fad8de54adda0e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 08:20:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d9f3538b446dadafb6eb2911359fe8233b3a7fa","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-22 08:20:55.000000000","message":"Removed Workflow+1 by Tobias Urdin \u003ctobias.urdin@binero.se\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"dd03d796ef4e79da86d3149940755dc6659c1be8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 08:41:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a35498386ab74c38b733ff2046261b96 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a7760500b054870b4f549d9a1c14c8a : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/4343da72edc141d892d342de6e7e1c16 : SUCCESS in 8m 30s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27936b565fde4a66b213445ecc8d40c9 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f908e68781224cb9aa9d60eb38e36723 : SUCCESS in 4m 32s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdec545fa4174721a31e99b1b50c009d : SUCCESS in 18m 55s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a224952ff07b43079c73b8bff9db2811 : FAILURE in 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b15149a50899ba7fb4999a1ed027c226a149bda","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-22 16:50:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9e065af233494594bcef8a0b62433bf2172f3a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 16:52:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc821d654c7135d94c817e06da4ba97feb6b0ec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:18:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e36a27bcbe5f4a2aa1506c3fab851bb5 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45de158727fd457691fb06ec5b59f116 : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/72d99e8e350548fcbbc159655fa91165 : SUCCESS in 10m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f295ff6d2604dc590c7f4ce4e147e55 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4850d17a092d4f96a6b3b64e9f4e7446 : SUCCESS in 3m 53s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ac2e7a206b6407ebe494585e1434104 : SUCCESS in 18m 39s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/615a042154b542309a11076401131db4 : FAILURE in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5fe8c3ecdbccef48f430e8fea9ae80ce0199bf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:18:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5bd4eef6b9678fff8ac8cfd407e710fabfc7a02","revisions":{"d5bd4eef6b9678fff8ac8cfd407e710fabfc7a02":{"kind":"REWORK","_number":1,"created":"2021-03-21 02:00:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/781979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/79/781979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/79/781979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/79/781979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/79/781979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/79/781979/1"}}},"commit":{"parents":[{"commit":"7df4d2cfbce20c35bb293b0bc5ae95542bc5c508","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/7df4d2cfbce20c35bb293b0bc5ae95542bc5c508"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-21 01:59:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-21 01:59:53.000000000","tz":540},"subject":"Add support for the oslo_policy/enforce_scope parameter","message":"Add support for the oslo_policy/enforce_scope parameter\n\nDepends-on: https://review.opendev.org/#/c/759008/\nChange-Id: Ia54c443e57539e8787b471fb627d1b1bd2ac0989\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/d5bd4eef6b9678fff8ac8cfd407e710fabfc7a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/d5bd4eef6b9678fff8ac8cfd407e710fabfc7a02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
