)]}'
{"id":"openstack%2Fpuppet-murano~789156","triplet_id":"openstack%2Fpuppet-murano~master~If0ecaecf1a0accccf161256292831255efc37159","project":"openstack/puppet-murano","branch":"master","topic":"setup-cfg","hashtags":[],"change_id":"If0ecaecf1a0accccf161256292831255efc37159","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-05-03 08:34:11.000000000","updated":"2021-05-11 09:33:34.000000000","submitted":"2021-05-11 09:33:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789156-setup-cfg","meta_rev_id":"617d56bf3e1ed48cb2561fc63114178d2ef44af3","_number":789156,"virtual_id_number":789156,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 09:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-05-11 09:19:32.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-05-03 14:17:11.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-05-11 09:19:32.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 08:39:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 14:17:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-11 09:19:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"f441a9628448db565c2645011e418694d4723279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-05-03 08:34:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1eef084d1f01ac11c232891b982e1b582de077c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 08:39:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d839d48d944ff59ed432b6abb8e33e : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a2b6ece3bb77d6befcdf52486ca30c236b279275","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-03 14:17:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"76b158618dbb6ca329c5d337c4551ae9fe12694c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 09:19:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72adc79bea3c2554cefa43b0c24e146b72fb93f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:19:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88781074db519459203b8567715885f1db9c27d4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 09:20:18.000000000","message":"Topic set to setup-cfg","accounts_in_message":[],"_revision_number":1},{"id":"ce44f8df14a5034063418c021224e0eaf2c0c250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:33:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d5a1f99dff54ee0a936f81e227f2c01 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"617d56bf3e1ed48cb2561fc63114178d2ef44af3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:33:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b26f87db1d80d34a290a621f7f6611f46418dad5","revisions":{"b26f87db1d80d34a290a621f7f6611f46418dad5":{"kind":"REWORK","_number":1,"created":"2021-05-03 08:34:11.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/56/789156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/56/789156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/56/789156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/56/789156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/56/789156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/56/789156/1"}}},"commit":{"parents":[{"commit":"2415ec0677b266829d450037bfdb7f987d9db72a","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/2415ec0677b266829d450037bfdb7f987d9db72a"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-05-03 08:33:28.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-05-03 08:34:07.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: If0ecaecf1a0accccf161256292831255efc37159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/b26f87db1d80d34a290a621f7f6611f46418dad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/b26f87db1d80d34a290a621f7f6611f46418dad5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
