)]}'
{"id":"openstack%2Fpuppet-murano~849920","triplet_id":"openstack%2Fpuppet-murano~master~Icc58f5c51b249564ab3cf3e1859c4dabde5857fc","project":"openstack/puppet-murano","branch":"master","topic":"fix-test-desc","hashtags":[],"change_id":"Icc58f5c51b249564ab3cf3e1859c4dabde5857fc","subject":"Fix wrong test description","status":"MERGED","created":"2022-07-15 02:48:45.000000000","updated":"2022-07-15 04:02:34.000000000","submitted":"2022-07-15 04:02:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849920-fix-test-desc","meta_rev_id":"8d5b307433164ea642e7a4c5d9042866b51f6625","_number":849920,"virtual_id_number":849920,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 04:02:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-07-15 03:40:18.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-07-15 03:40:18.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 03:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b4c5bc38a6f39247321661ef88840de447c4b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-15 02:48:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c64f100948d1758f6c6a634702e8d73cf59f0569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 03:16:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfba43b0825e457aa7d6c551f5bceff3\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8cbcdde0b2248f6b19a2d0ec8a77ce2 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ae83d0ff10d44b58c10019ba7793edb : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a09666cdd66749c3b1c0d1860c15b99b : SUCCESS in 3m 01s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca09012475dc4d49925a6518928dfed9 : SUCCESS in 11m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e49f2a650f134445a2e3e971012fdf9d : SUCCESS in 9m 55s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c6aead651af45938e8ff96848e554a1 : SUCCESS in 9m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cd128ab800a41eaa740eb39401c8872 : SUCCESS in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e4c6669fafed46854e7a2295b15ae0ed1727c7d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-15 03:40:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66b33406e4d94cc31071fe5d998f57e1a235ad95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 03:41:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a11920cd9b5ba2e1f9bd83ed516f58c64fea129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 04:02:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abdd1fb3dba7448f94f3fc132b602f88\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c642ca397f84b14a5b4468b02602335 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4da06614fafd4391982d83b104215d47 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48764987e3dd4d3d969af678a4de683f : SUCCESS in 5m 29s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/131873a2f7ac4d5290cb739f705eafed : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfe1743d53384636a5997130354500f3 : SUCCESS in 10m 43s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/454f3f25b3ef462e8cf09192abcea2dd : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8d5b307433164ea642e7a4c5d9042866b51f6625","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 04:02:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc7467334efbc95d6c7bafd569304980477338a1","revisions":{"fc7467334efbc95d6c7bafd569304980477338a1":{"kind":"REWORK","_number":1,"created":"2022-07-15 02:48:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/849920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/20/849920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/20/849920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/20/849920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/20/849920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/20/849920/1"}}},"commit":{"parents":[{"commit":"aa9b0aac05b1e3c9f59e1c609fb293a5e2579141","subject":"Fix some typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/aa9b0aac05b1e3c9f59e1c609fb293a5e2579141"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-15 02:48:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-15 02:48:36.000000000","tz":540},"subject":"Fix wrong test description","message":"Fix wrong test description\n\nThe value with a white space should be accepted instead of being\nrejected. This change fixes the wrong test case description and makes\nsure the description explains what is actually tested.\n\nChange-Id: Icc58f5c51b249564ab3cf3e1859c4dabde5857fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/fc7467334efbc95d6c7bafd569304980477338a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/fc7467334efbc95d6c7bafd569304980477338a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
