)]}'
{"id":"openstack%2Fpuppet-murano~870449","triplet_id":"openstack%2Fpuppet-murano~master~I7a6e52b1c306c048eed09f904f447f68ca205f8e","project":"openstack/puppet-murano","branch":"master","topic":"puppet-jammy","hashtags":[],"change_id":"I7a6e52b1c306c048eed09f904f447f68ca205f8e","subject":"Switch to Ubuntu Jammy (22.04)","status":"MERGED","created":"2023-01-15 13:25:38.000000000","updated":"2023-01-16 00:38:41.000000000","submitted":"2023-01-16 00:37:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870449-puppet-jammy","meta_rev_id":"5083707750b2d88c3c3236406c050f315f45db85","_number":870449,"virtual_id_number":870449,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 00:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-01-16 00:23:51.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-01-16 00:23:51.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-15 13:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d7cafd84a10e81c4aa5746d4ce4281fbe15fd8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-15 13:25:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41018fca175d82631105e438f614f62a601167de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 13:37:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e594bef7e8242a3821246c75b3cbb8f\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ea8b16237c5a4d07be2941966e37e504 : SUCCESS in 1m 51s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3153ba4668449e88807ac12942c780f : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae7053c50461401ea09a72f23649e967 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4543724fd4f540ee101e84281e1405d019cfe042","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-16 00:23:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10ba64dc45b9daecf520f969a9262f43b5133344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:24:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"875a2320e7b4c0ab7432ef7e25298dbae448bad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:37:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476fadd9bf434677b4f0dc53cae96a55\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/7cce30607aa041ecb19f504a6fe89540 : SUCCESS in 1m 58s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb5b997ca3e84d128a7abbc5bec886d4 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67805d3c51fa4e29bbf9612eeffaccc2 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"69b6002c3e116234175d3efdef6833ecdf818701","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:37:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5083707750b2d88c3c3236406c050f315f45db85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:38:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13d5dd33bcba486290892d083dc9c9fc\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b99165813c2490a9e23b1d677b96b4a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4890cfc26c502c1cc863b1440012cc7c513b23b","revisions":{"e4890cfc26c502c1cc863b1440012cc7c513b23b":{"kind":"REWORK","_number":1,"created":"2023-01-15 13:25:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/870449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/49/870449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/49/870449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/49/870449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/49/870449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/49/870449/1"}}},"commit":{"parents":[{"commit":"0a53e1d4ac6296bbaab5704dc87d34db90bf54fa","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/0a53e1d4ac6296bbaab5704dc87d34db90bf54fa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-15 13:25:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-15 13:25:30.000000000","tz":540},"subject":"Switch to Ubuntu Jammy (22.04)","message":"Switch to Ubuntu Jammy (22.04)\n\n... because Focal no longer supports the recent releases such as Zed.\n\nChange-Id: I7a6e52b1c306c048eed09f904f447f68ca205f8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/e4890cfc26c502c1cc863b1440012cc7c513b23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/e4890cfc26c502c1cc863b1440012cc7c513b23b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
