)]}'
{"id":"openstack%2Fpuppet-murano~871451","triplet_id":"openstack%2Fpuppet-murano~master~Idf2da530a4fefabae350ff0d67a5c8770ab22ed9","project":"openstack/puppet-murano","branch":"master","topic":"auth-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-24 08:53:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Idf2da530a4fefabae350ff0d67a5c8770ab22ed9","subject":"Remove deprecated authtoken parameters of the base class","status":"MERGED","created":"2023-01-23 08:31:10.000000000","updated":"2023-01-24 10:01:02.000000000","submitted":"2023-01-24 09:59:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871451-auth-opts","meta_rev_id":"9caf5c78f1cd40dacf0103e096270c784022d25c","_number":871451,"virtual_id_number":871451,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 09:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-01-23 17:58:15.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-01-23 19:36:10.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-01-23 19:36:10.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 09:01:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 17:58:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-01-23 19:36:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"6688b745d9d5678a4ce7593e18c42164b445c76a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 08:31:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"572ab8ea4608aa8b79cbbec27c1438af2bef39f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 09:01:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cadbe901828c4f5ea65117f6d334534c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87947f4b62c34d05bf0c7d228fa04394 : FAILURE in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a37a9b748bc04d9ab71f38a29647d365 : SUCCESS in 2m 24s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd00c218f01d49c3b01af4552bf2c4e9 : SUCCESS in 13m 10s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35970869b3c240898919b31684261b40 : SUCCESS in 7m 24s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30d0ed3a202b4c799eb3d24d7b2bb2e5 : SUCCESS in 7m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e49cc2a2f4e84420a0ebd40044eabb41 : SUCCESS in 3m 45s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0130643a23914d2ea22fbaa7e10e24e8 : SUCCESS in 17m 53s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/855eb38a620742f4befc35b82e16c6ab : FAILURE in 6m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0faf6f57eca04f81b747ae43ad179535 : FAILURE in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78bf4445f45703ed01cc52758ce29037b5948649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 09:05:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6ca0d071c0521c737c052f2065cb6efa3e9c410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 09:22:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df4b062f3b104ddca2b8ca633f513ce4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed4fb58cda03442db15321ff83abf3e0 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf1abea24a864b6a96b49a0d67d82205 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/864aae57ff8c44b19d8afbe362bb8e3e : SUCCESS in 8m 49s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44866cef38344eb79567ff9dfb12c978 : SUCCESS in 6m 24s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b4d236d35a94a2aab353096c8c97f6c : SUCCESS in 7m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e392ea50f534b71be6b53c1b4082e0f : SUCCESS in 2m 57s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5763bfdac354eadaeaeb68179487073 : SUCCESS in 14m 33s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bcc5df383d14b5ea62d178fe4e340f6 : FAILURE in 6m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/716140e1c07a4b1c900c6cb4d8a8538c : FAILURE in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"607c07c0da7920e36cf3b0a86ad6bba00154b2d7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-23 17:58:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b179f687a560408e930b25ec67d97ba20aadf54","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-23 19:36:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"76716cd4e22f8b44be962104e0698b1770e65cee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:38:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"54747582f8363328c74ab91674b53ea5fb9c77b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 20:03:38.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/871453 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"3e14cc6651f5b6d31b8f4dd95bc2b9399ddf9829","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-24 08:53:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e46645968a861b2e810e2ece2fa1fdc1c343f64c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 09:33:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a43897a1a3bc4ce5be00243725021975\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a18617f2c0fb47e8af7e279c18c592bb : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a73107f9c2054c15a94772a5fc17fb6d : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/83f23fe05c564572a86579e6088da02f : SUCCESS in 9m 22s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af48a7949f0e420593d8ebc0a763d23a : SUCCESS in 6m 43s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4246de99ebe4f10b9d13cde95d4808e : SUCCESS in 8m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42be82c63c234650877a9899c8a5b91c : SUCCESS in 3m 03s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/79294b453b9d446da72cc299fb4f36dd : SUCCESS in 17m 55s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39f179812b714f978129894e8a432b08 : FAILURE in 10m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d53d4c6700134751b1418b0f15eabd08 : FAILURE in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b20e49f25f49b420af9c1ce0bc36c4ad5c19bcfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 09:38:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"55189d4ec0edc0e0218f7b3b88178b0a1f485e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 09:59:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/341aedc2ea15412f89d71a97c3d7b90d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1ab2df01dd24922ae43032cda062f05 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84785adaebe04f6c9dc6a9b00d0e36d7 : SUCCESS in 2m 23s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6017735ae2be408d86d9dd90acd2192d : SUCCESS in 10m 43s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e6307e572cb4946b6965039f3340c1d : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a880760d733f4765a16cd7829ef926ef : SUCCESS in 2m 46s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/95e7dfd313f744d98ef16b539dcb9fab : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":2},{"id":"b725f6da328bd7c9d69a0981f90fbf081db61721","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 09:59:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9caf5c78f1cd40dacf0103e096270c784022d25c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:01:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88599eac123749549d5c6cc261a23717\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6653cbe8cc08412b924a01c0416e56a6 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"859ea03fe923043e672de0e44518e3f9424ec9db","revisions":{"6611a87e0618f07d009dd7771be0895667a99325":{"kind":"REWORK","_number":1,"created":"2023-01-23 08:31:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/871451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/51/871451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/51/871451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/51/871451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/51/871451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/51/871451/1"}}},"commit":{"parents":[{"commit":"e4890cfc26c502c1cc863b1440012cc7c513b23b","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/e4890cfc26c502c1cc863b1440012cc7c513b23b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 08:29:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 08:30:51.000000000","tz":540},"subject":"Remove deprecated authtoken parameters of the base class","message":"Remove deprecated authtoken parameters of the base class\n\nThese parameters were deprecated in favor of the separate class added\nduring Yoga cycle[1].\n\n[1] 183e14ab35c7cf379546b235ea98983c1c9d0259\n\nChange-Id: Idf2da530a4fefabae350ff0d67a5c8770ab22ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/6611a87e0618f07d009dd7771be0895667a99325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/6611a87e0618f07d009dd7771be0895667a99325"}]},"branch":"refs/heads/master"},"859ea03fe923043e672de0e44518e3f9424ec9db":{"kind":"REWORK","_number":2,"created":"2023-01-23 09:05:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/871451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/51/871451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/51/871451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/51/871451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/51/871451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/51/871451/2"}}},"commit":{"parents":[{"commit":"153dcef0fae2b2fca205146788d34211f1812868","subject":"Do not validate optional admin_password","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/153dcef0fae2b2fca205146788d34211f1812868"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 08:29:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 09:05:23.000000000","tz":540},"subject":"Remove deprecated authtoken parameters of the base class","message":"Remove deprecated authtoken parameters of the base class\n\nThese parameters were deprecated in favor of the separate class added\nduring Yoga cycle[1].\n\n[1] 183e14ab35c7cf379546b235ea98983c1c9d0259\n\nChange-Id: Idf2da530a4fefabae350ff0d67a5c8770ab22ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/859ea03fe923043e672de0e44518e3f9424ec9db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/859ea03fe923043e672de0e44518e3f9424ec9db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
