)]}'
{"id":"openstack%2Fpuppet-murano~873878","triplet_id":"openstack%2Fpuppet-murano~master~I615ad890e650eeaa8211188447c60bd5fa336fb2","project":"openstack/puppet-murano","branch":"master","hashtags":[],"change_id":"I615ad890e650eeaa8211188447c60bd5fa336fb2","subject":"Testing litmus job","status":"ABANDONED","created":"2023-02-15 08:14:57.000000000","updated":"2023-02-15 09:04:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4f72cba2d96015286c42ea32bfa60961ab4a20bb","_number":873878,"virtual_id_number":873878,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-15 08:40:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 08:40:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c86d75ab95f50e2a47d3e72dcece941b31e90ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-15 08:14:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e5cbe3bcc210af5dd9d54367953cfddff2b1a4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 08:40:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/515e7c60a77b4949887f9d10f601b514\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b374e3f338f41c8b24ee2501500d035 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9223b8ab562f4d4d8d9596a39e3e62da : SUCCESS in 2m 51s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/41188750ab844f9b8f38286c8de7be8f : SUCCESS in 20m 22s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c52c3576bdfc41269626a597918bb94e : FAILURE in 14m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e925f64e3e5847cc9a98e0544b453360 : FAILURE in 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f72cba2d96015286c42ea32bfa60961ab4a20bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-15 09:04:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"653e14d7cbb801ad7e73c7f6296d35ac550877ab","revisions":{"653e14d7cbb801ad7e73c7f6296d35ac550877ab":{"kind":"REWORK","_number":1,"created":"2023-02-15 08:14:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/873878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/78/873878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/873878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/873878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/873878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/78/873878/1"}}},"commit":{"parents":[{"commit":"fdd971ec5d03d7a300decd376d9d7b31f2c1da70","subject":"Merge \"CentOS: Enable murano in acceptance tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/fdd971ec5d03d7a300decd376d9d7b31f2c1da70"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-15 08:14:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-15 08:14:35.000000000","tz":540},"subject":"Testing litmus job","message":"Testing litmus job\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/873877\nChange-Id: I615ad890e650eeaa8211188447c60bd5fa336fb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/653e14d7cbb801ad7e73c7f6296d35ac550877ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/653e14d7cbb801ad7e73c7f6296d35ac550877ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
