)]}'
{"id":"openstack%2Fpuppet-murano~894478","triplet_id":"openstack%2Fpuppet-murano~master~Id8970f67c8b9dc7cbf4026db28f5715f7e025fb7","project":"openstack/puppet-murano","branch":"master","topic":"policy-reload","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-11 04:14:43.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Id8970f67c8b9dc7cbf4026db28f5715f7e025fb7","subject":"Do not restart services after policy file changes","status":"MERGED","created":"2023-09-11 03:30:38.000000000","updated":"2023-09-11 05:32:44.000000000","submitted":"2023-09-11 05:32:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894478-policy-reload","meta_rev_id":"eeac1d93fb6fc4e99e676779b6eae340858b23ba","_number":894478,"virtual_id_number":894478,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-11 05:32:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-09-11 05:01:01.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-09-11 05:01:01.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 04:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e923d7eae3a4c6ca38da3d1d5859dc2268039f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 03:30:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1078bddee345390b03251996ef2add145fadc1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 04:02:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f82dba7be9f4166ac92fb080dcf762b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38ecaa9dbb1d46d7833c2ab40c524781 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b199519dd6cd46a39e88e4ae870fd9e2 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb3d489d195940889802b58b8c48cd15 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/57706c9460aa46f087b4ec0624b4b17e : SUCCESS in 12m 11s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dca3ac6b481d4c67ba5e8bda632d9e13 : SUCCESS in 9m 36s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd3ef31d0a674cc794d307e8b1acb5ba : FAILURE in 9m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9f1c5811ac3403a931796588ac9ae4a : SUCCESS in 6m 42s (non-voting)\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/55d6cc9fef4d4db2bec272e87c2ded01 : FAILURE in 27m 41s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62c57f741bff48da8cd67af079c961bb : FAILURE in 20m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7c5b4ed30ad47a2a9e64c7d645ef091 : FAILURE in 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26cd02aee798ebfd2b5b13f383e131ba34d10aeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 04:13:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"700223ff31243ade56464337ac44245e30de61e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 04:25:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"faa4ccf6295a0f758bb9b45e86c6db9e7c8943f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 04:42:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123a966a0df9479bad7781b5530cbebe\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae5804e3d672440bb6a73a898fec046f : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59ecd6a5da7046978c3b962cbc8f0a9a : SUCCESS in 2m 40s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ec2e9abb9384c5f9082a8ae24cc5176 : SUCCESS in 5m 45s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/870ff79fa7c84d7d8a04f487a3eb94db : SUCCESS in 8m 43s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18f09adfa7f74f1c9b17d22a58bc237f : SUCCESS in 6m 19s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/5684e7e157314383a38323dd2569f59a : FAILURE in 11m 51s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e382134665e4d249c6045094f4358d9 : SUCCESS in 6m 41s (non-voting)\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/374ce1b403e242eba904ed0a075e261b : SUCCESS in 15m 19s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc75b24d627042e896df3cdf80a93e50 : FAILURE in 15m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab7df25ba1ba49ef8f974d6994643875 : FAILURE in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7bfcccc8667f12e6545405747ffb0506d049c6cd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 05:01:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eeb929d2ce611a06ad8f8cf06ce5ae0678c42429","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:10:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9b144417c617126e45f1288aa45d653d94b36c69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:32:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27ab8beaba204a9e95f7797104688bc3\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/942b80bd8b4543eda0fe6fd62b7bb49d : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d7231eb95f6491c9814ea084499327e : SUCCESS in 2m 51s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/919db1cbdce1426aa728dc26d0b532ca : SUCCESS in 9m 24s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ead5f9fc27414b4cafedecc79dbc3f5d : SUCCESS in 6m 30s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/2965cb2f7c574f3187122bfbd82f0d18 : FAILURE in 9m 55s (non-voting)\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b2277dfeeb44da69b1d0b37bdb9944d : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":3},{"id":"eeac1d93fb6fc4e99e676779b6eae340858b23ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:32:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ea95a1b182a30aa1ca778679b799c7cc4703675f","revisions":{"ca2642310613d1466ddadc87782696e87d5a86d4":{"kind":"REWORK","_number":1,"created":"2023-09-11 03:30:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/894478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/78/894478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/1"}}},"commit":{"parents":[{"commit":"c2e58e36d342b6c7a09828a068ba6408c7e246cb","subject":"Exclude release note files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/c2e58e36d342b6c7a09828a068ba6408c7e246cb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 03:30:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 03:30:29.000000000","tz":540},"subject":"Do not restart services after policy file changes","message":"Do not restart services after policy file changes\n\nThe oslo.policy library has implementations to detect change in policy\nrules and reload the new rules without service restart.\n\nChange-Id: Id8970f67c8b9dc7cbf4026db28f5715f7e025fb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/ca2642310613d1466ddadc87782696e87d5a86d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/ca2642310613d1466ddadc87782696e87d5a86d4"}]},"branch":"refs/heads/master"},"06ef2e30fc5e6fde04b3f185420cab673909c206":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-11 04:13:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/894478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/78/894478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/2"}}},"commit":{"parents":[{"commit":"24e01e24c21c9cd977bfbc0c9d6fa0a9309b741d","subject":"Enable memcached in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/24e01e24c21c9cd977bfbc0c9d6fa0a9309b741d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 03:30:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 04:13:27.000000000","tz":540},"subject":"Do not restart services after policy file changes","message":"Do not restart services after policy file changes\n\nThe oslo.policy library has implementations to detect change in policy\nrules and reload the new rules without service restart.\n\nChange-Id: Id8970f67c8b9dc7cbf4026db28f5715f7e025fb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/06ef2e30fc5e6fde04b3f185420cab673909c206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/06ef2e30fc5e6fde04b3f185420cab673909c206"}]},"branch":"refs/heads/master"},"ea95a1b182a30aa1ca778679b799c7cc4703675f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-11 04:25:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/894478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/78/894478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/78/894478/3"}}},"commit":{"parents":[{"commit":"840927a91c49db99183806907d9c5cbadca0d16e","subject":"Enable memcached in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/840927a91c49db99183806907d9c5cbadca0d16e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 03:30:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 04:25:23.000000000","tz":0},"subject":"Do not restart services after policy file changes","message":"Do not restart services after policy file changes\n\nThe oslo.policy library has implementations to detect change in policy\nrules and reload the new rules without service restart.\n\nChange-Id: Id8970f67c8b9dc7cbf4026db28f5715f7e025fb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/ea95a1b182a30aa1ca778679b799c7cc4703675f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/ea95a1b182a30aa1ca778679b799c7cc4703675f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
