)]}'
{"id":"openstack%2Fpuppet-murano~901414","triplet_id":"openstack%2Fpuppet-murano~master~I83483123e4d900112ba409c282d7607ed382e31c","project":"openstack/puppet-murano","branch":"master","topic":"client-package-tag","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-20 17:45:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83483123e4d900112ba409c282d7607ed382e31c","subject":"Use new openstackclient tag","status":"MERGED","created":"2023-11-19 16:24:33.000000000","updated":"2023-11-20 17:45:42.000000000","submitted":"2023-11-20 17:45:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901414-client-package-tag","meta_rev_id":"9d24b09c9a85ae667fa31b931d739fa7d47364e9","_number":901414,"virtual_id_number":901414,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-20 17:45:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-20 09:17:47.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-20 17:17:54.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-20 17:17:54.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-19 16:47:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-20 09:17:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-20 17:17:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ececb599ab07a71e1bcbc5936a41d4502c90524a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-19 16:24:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f012aa71d2f94a24a3644693e945e0ed148a2835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-19 16:25:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"415eaeb0ec38b908f1edbbc08c055fc3343061e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-19 16:26:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"95c34e70833e71c12487691cee846fbe18d0f97c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-19 16:26:36.000000000","message":"Topic set to client-package-tag","accounts_in_message":[],"_revision_number":3},{"id":"20adb8fd01088c6e5282ee3c4378fbe9f9de9b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 16:47:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c2d5b3e949844cdb3328f88e01c7f21\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aba73347a8374e5286841d34a47b51bb : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/005e1ca54d8f433ba5ee829d0009f972 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a1416daea474f74af41dfcb57a8b629 : SUCCESS in 4m 48s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb560b035746436ca4fa9e8c83828f56 : SUCCESS in 8m 04s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6481381a7454e2186f861b9c20425f4 : SUCCESS in 6m 13s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/98ffe10f920d48ec9c2c184aa184b267 : SUCCESS in 8m 15s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba8da05749a9487d9f81584f96bd13b6 : SUCCESS in 5m 35s (non-voting)\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/389d0ce3390549e392d6b92e33cd41ff : FAILURE in 17m 58s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1602a6dfa4b74031b28f572ff55c6871 : SUCCESS in 11m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c45235961564475bbf9b34ada313e02 : SUCCESS in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9dbdf22a664de73a6a8bcd25deb5dd59a4c714fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-19 17:03:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a01fb596c391b728880ccc3aaad0ab62389f8b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 17:26:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2b5635c7b1f4b0fb0127ba373602852\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cee7d859cab4f5f812cf067f8402bfa : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad132094714b404899b4370aeb31b8b9 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b122655fd0f041d7ba8d63926d9fcb88 : SUCCESS in 5m 09s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d2d1781ed1c49a7b2f4400deab60341 : SUCCESS in 9m 27s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/653991d984eb4af280cf983e995d944d : SUCCESS in 5m 35s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/34d34ce61a3940868f100a30b3499edd : SUCCESS in 7m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d77147c8bba4eef96e7c13be2988d39 : SUCCESS in 5m 35s (non-voting)\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/9472bcab31b8452bad4fc21a677f1da4 : SUCCESS in 17m 30s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eae27757d00346f3809b676a4e7be6ae : SUCCESS in 11m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d5e5ca6ece8406988145635b00135d7 : SUCCESS in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75d06f8f426a7bef4454c6b296a927907d2f8d88","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-20 09:17:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7289d2c437d6c9a655faa46c1f7aefe7b8f14bc7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-20 17:17:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b2f6245cc4e3ad8b28690ac0b4dabc2d07e2f04c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 17:18:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"913ff73ed6db6dd10b90a428fb60d798cf652747","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 17:45:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee03d02bdee4d448a439598f6ba09c2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc1c48b6f4624583a4ce5352fff9ef2a : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d545bdbe5050415981edb2c19033b0c5 : SUCCESS in 2m 45s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/50c3d8dd54f248cfb6f02bb4059e7c03 : SUCCESS in 9m 38s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40b641edd1ad49c6a968d85107251ffe : SUCCESS in 7m 32s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/74e5b738d38e4020a502f7baf5a01abf : SUCCESS in 10m 18s (non-voting)\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdae6d478b1640839f7dfc07ef939eb0 : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":4},{"id":"9d24b09c9a85ae667fa31b931d739fa7d47364e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 17:45:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"520b8d3ecbaae3c0cdf46a27bf6e92accf223aac","revisions":{"fd1616906b3387013c2e0faf1d93d54c65d50428":{"kind":"REWORK","_number":1,"created":"2023-11-19 16:24:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/901414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/14/901414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/1"}}},"commit":{"parents":[{"commit":"b5b3082b18f43a227ce83f4e6cac271f65e8778d","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/b5b3082b18f43a227ce83f4e6cac271f65e8778d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-19 16:24:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-19 16:24:21.000000000","tz":540},"subject":"Use new openstackclient tag","message":"Use new openstackclient tag\n\nThis looses dependency using the new openstackclient tag, which\nrequires only packages actually related to openstack CLI.\n\nDepends-on: https://review.opendev.org/899594\nChange-Id: I83483123e4d900112ba409c282d7607ed382e31c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/fd1616906b3387013c2e0faf1d93d54c65d50428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/fd1616906b3387013c2e0faf1d93d54c65d50428"}]},"branch":"refs/heads/master"},"3e7473a6e990058cadc9e679dd67b7c87767a4c2":{"kind":"REWORK","_number":2,"created":"2023-11-19 16:25:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/901414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/14/901414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/2"}}},"commit":{"parents":[{"commit":"b5b3082b18f43a227ce83f4e6cac271f65e8778d","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/b5b3082b18f43a227ce83f4e6cac271f65e8778d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-19 16:24:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-19 16:24:55.000000000","tz":540},"subject":"Use new openstackclient tag","message":"Use new openstackclient tag\n\nThis looses dependency using the new openstackclient tag, which\nrequires only packages actually related to openstack CLI.\n\nThis also fixes the existing wrong tag (murano-package\u0027s\u0027 instead of\nmurano-package).\n\nDepends-on: https://review.opendev.org/899594\nChange-Id: I83483123e4d900112ba409c282d7607ed382e31c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/3e7473a6e990058cadc9e679dd67b7c87767a4c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/3e7473a6e990058cadc9e679dd67b7c87767a4c2"}]},"branch":"refs/heads/master"},"f74264b15c1cef8a645f1b65be4c3377ecf0b2aa":{"kind":"REWORK","_number":3,"created":"2023-11-19 16:26:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/901414/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/14/901414/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/3"}}},"commit":{"parents":[{"commit":"b5b3082b18f43a227ce83f4e6cac271f65e8778d","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/b5b3082b18f43a227ce83f4e6cac271f65e8778d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-19 16:24:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-19 16:26:08.000000000","tz":540},"subject":"Use new openstackclient tag","message":"Use new openstackclient tag\n\nThis looses dependency using the new openstackclient tag, which\nrequires only packages actually related to openstack CLI.\n\nThis also fixes the existing wrong tag (murano-package\u0027s\u0027 instead of\nmurano-package).\n\nDepends-on: https://review.opendev.org/899594\nChange-Id: I83483123e4d900112ba409c282d7607ed382e31c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/f74264b15c1cef8a645f1b65be4c3377ecf0b2aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/f74264b15c1cef8a645f1b65be4c3377ecf0b2aa"}]},"branch":"refs/heads/master"},"520b8d3ecbaae3c0cdf46a27bf6e92accf223aac":{"kind":"REWORK","_number":4,"created":"2023-11-19 17:03:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/901414/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-murano","ref":"refs/changes/14/901414/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-murano refs/changes/14/901414/4"}}},"commit":{"parents":[{"commit":"b5b3082b18f43a227ce83f4e6cac271f65e8778d","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/b5b3082b18f43a227ce83f4e6cac271f65e8778d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-19 16:24:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-19 17:03:10.000000000","tz":540},"subject":"Use new openstackclient tag","message":"Use new openstackclient tag\n\nThis looses dependency using the new openstackclient tag, which\nrequires only packages actually related to openstack CLI.\n\nThis also fixes the existing wrong tag (murano-package\u0027s\u0027 instead of\nmurano-package).\n\nDepends-on: https://review.opendev.org/899594\nChange-Id: I83483123e4d900112ba409c282d7607ed382e31c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/520b8d3ecbaae3c0cdf46a27bf6e92accf223aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-murano/commit/520b8d3ecbaae3c0cdf46a27bf6e92accf223aac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
