)]}'
{"id":"openstack%2Fpuppet-neutron~655601","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fqueens~I7c52d29e4f147d84c4e34bf7fdd067df827c7949","project":"openstack/puppet-neutron","branch":"stable/queens","topic":"bug/1781011-stable/queens","hashtags":[],"change_id":"I7c52d29e4f147d84c4e34bf7fdd067df827c7949","subject":"Fix dependency cycle in networking-ansible plugin manifest","status":"MERGED","created":"2019-04-25 02:59:24.000000000","updated":"2019-04-25 15:20:04.000000000","submitted":"2019-04-25 15:20:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655601-1556205604555-af911147","meta_rev_id":"ac07b30b9befea92d676de8dd942d2522bf6e71d","_number":655601,"virtual_id_number":655601,"owner":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-25 11:46:40.000000000","_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-25 15:20:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":0,"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"value":1,"date":"2019-04-25 11:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-04-25 11:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"date":"2019-04-25 11:37:53.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-25 14:07:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 11:37:31.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2019-04-25 11:37:53.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-25 11:44:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-25 14:07:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-25 15:20:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b313221c8373b828e34b05d5d4943d7c62a7e145","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2019-04-25 02:59:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"25a4429ca954d2fd1660cd7102886cd8495aa937","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 04:12:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/01/655601/1/check/puppet-openstack-lint/05c248d/ : SUCCESS in 6m 30s\n- puppet-openstack-syntax-4 http://logs.openstack.org/01/655601/1/check/puppet-openstack-syntax-4/41a10b0/ : SUCCESS in 3m 51s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/01/655601/1/check/puppet-openstack-unit-4.8-centos-7/bd2aa1b/ : SUCCESS in 26m 22s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/655601/1/check/puppet-openstack-unit-5.5-centos-7/b6931d0/ : SUCCESS in 18m 52s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous http://logs.openstack.org/01/655601/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous/da28ca0/ : SUCCESS in 1h 04m 22s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/01/655601/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/8fbae13/ : SUCCESS in 1h 01m 43s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/01/655601/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/becaf4e/ : SUCCESS in 1h 02m 33s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous http://logs.openstack.org/01/655601/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous/844e345/ : SUCCESS in 57m 22s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous http://logs.openstack.org/01/655601/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous/13935c8/ : SUCCESS in 42m 07s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/01/655601/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/6836a40/ : SUCCESS in 42m 47s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/01/655601/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/4651e13/ : SUCCESS in 35m 06s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous http://logs.openstack.org/01/655601/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous/3625ea8/ : SUCCESS in 33m 32s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/01/655601/1/check/puppet-openstack-beaker-centos-7/e309cd5/ : SUCCESS in 16m 10s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/01/655601/1/check/puppet-openstack-beaker-xenial/ac7a6a1/ : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e19c6aea6fc2800f4a5290f7af35236c150e8f52","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-04-25 11:37:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"19c17f99edc34266eee3c944ddb768cb20eea162","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-25 11:44:47.000000000","message":"Patch Set 1: Code-Review+2\n\nMaybe backport the changed in master to the other stable branches as well if it\u0027s broken there so it\u0027s not only fixed in master and queens.","accounts_in_message":[],"_revision_number":1},{"id":"7f31c47d0efab729f0c26c5c10a7c7ed57aebd97","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2019-04-25 11:46:40.000000000","message":"Patch Set 1:\n\nIt was only necessary in queens, I checked everything else newer already.","accounts_in_message":[],"_revision_number":1},{"id":"469125066f19a82c99e3dd5514a5d66f5bdec079","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-04-25 11:50:07.000000000","message":"Patch Set 1:\n\n\u003e Maybe backport the changed in master to the other stable branches as well if it\u0027s broken\n\nYeah, it was fixed in puppet-neutron 13.3.0 [1](rocky), so no more stable branches to backport.\n\n[1] https://bugs.launchpad.net/puppet-neutron/+bug/1781011/comments/3","accounts_in_message":[],"_revision_number":1},{"id":"bcd6198b88344041e410009dc4025a03aab471de","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-25 14:07:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b14285bd4f19394ae5929f1329cb55aa32e393ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:07:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2a4bd556cf153c2e0f436c46d56b6f59521afb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:20:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/01/655601/1/gate/puppet-openstack-lint/0b6c985/ : SUCCESS in 4m 24s\n- puppet-openstack-syntax-4 http://logs.openstack.org/01/655601/1/gate/puppet-openstack-syntax-4/67e251e/ : SUCCESS in 3m 50s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/01/655601/1/gate/puppet-openstack-unit-4.8-centos-7/e43397f/ : SUCCESS in 29m 01s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/655601/1/gate/puppet-openstack-unit-5.5-centos-7/89c7b0c/ : SUCCESS in 25m 31s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous http://logs.openstack.org/01/655601/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous/bf5286a/ : SUCCESS in 1h 01m 34s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/01/655601/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/f49de02/ : SUCCESS in 1h 05m 29s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/01/655601/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/9510201/ : SUCCESS in 1h 06m 21s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous http://logs.openstack.org/01/655601/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous/c4b13e7/ : SUCCESS in 53m 57s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous http://logs.openstack.org/01/655601/1/gate/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous/14df1a1/ : SUCCESS in 43m 24s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/01/655601/1/gate/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/2fa4c42/ : SUCCESS in 41m 19s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/01/655601/1/gate/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/5040319/ : SUCCESS in 36m 03s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous http://logs.openstack.org/01/655601/1/gate/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous/b25141a/ : SUCCESS in 35m 00s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/01/655601/1/gate/puppet-openstack-beaker-centos-7/db46f4d/ : SUCCESS in 19m 08s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/01/655601/1/gate/puppet-openstack-beaker-xenial/c2aa371/ : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9acb2d8674435b7ee466c53b846352a6e11b5f58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:20:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6a5bd82b59494b97c985fd02eae7f7f69311de5","revisions":{"a6a5bd82b59494b97c985fd02eae7f7f69311de5":{"kind":"REWORK","_number":1,"created":"2019-04-25 02:59:24.000000000","uploader":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"ref":"refs/changes/01/655601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/655601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/655601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/655601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/655601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/655601/1"}}},"commit":{"parents":[{"commit":"e6a6f42c2681d7abeb4f444743c9184364dcb17e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e6a6f42c2681d7abeb4f444743c9184364dcb17e"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2018-07-10 16:28:47.000000000","tz":330},"committer":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2019-04-25 02:59:24.000000000","tz":0},"subject":"Fix dependency cycle in networking-ansible plugin manifest","message":"Fix dependency cycle in networking-ansible plugin manifest\n\nRemoves the redundant dependency from networking_ansible.pp.\n\nChange-Id: I7c52d29e4f147d84c4e34bf7fdd067df827c7949\nCloses-Bug: #1781011\n(cherry picked from commit d6c0500b086ab7500a3a50c245d4696a49d93054)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a6a5bd82b59494b97c985fd02eae7f7f69311de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a6a5bd82b59494b97c985fd02eae7f7f69311de5"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
