)]}'
{"id":"openstack%2Fpuppet-neutron~659414","triplet_id":"openstack%2Fpuppet-neutron~master~I857de092429e94890064a16475c0442f3751ae45","project":"openstack/puppet-neutron","branch":"master","topic":"add_openstackclient","hashtags":[],"change_id":"I857de092429e94890064a16475c0442f3751ae45","subject":"Add openstackclient installation to the client class","status":"MERGED","created":"2019-05-16 01:53:16.000000000","updated":"2019-05-20 09:39:26.000000000","submitted":"2019-05-20 09:39:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659414-1558345166666-a663f06a","meta_rev_id":"363ec424e2c92a3a7a2f19208a5bc14fd0a2da0c","_number":659414,"virtual_id_number":659414,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-20 09:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-20 04:49:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-17 19:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-05-17 19:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-20 01:04:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 19:27:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-05-17 19:27:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-20 04:49:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-20 09:39:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85f6ea18cc5164ad52b9bd2d3dcc556aa5b472c9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-16 01:53:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be6d72673fd2e41844a0a7dc13ebb8e127c8e3ba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 03:34:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/14/659414/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/4761b8b/ : SUCCESS in 1h 34m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/14/659414/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/3df642f/ : SUCCESS in 1h 39m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/14/659414/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/7bd8f5d/ : SUCCESS in 1h 37m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/14/659414/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/6e9d389/ : FAILURE in 1h 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e93d7851aaaebcbc61bf29f92c0f3bc46fa0ee0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:04:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-lint-ubuntu-bionic/c256bd0/ : SUCCESS in 6m 24s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-syntax-4-ubuntu-bionic/edae90b/ : SUCCESS in 6m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-syntax-5-ubuntu-bionic/af86a46/ : SUCCESS in 6m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-syntax-6-ubuntu-bionic/aece3a7/ : SUCCESS in 6m 23s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/14/659414/1/check/puppet-openstack-unit-4.8-centos-7/9c7b7ba/ : SUCCESS in 27m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/14/659414/1/check/puppet-openstack-unit-5.5-centos-7/8d4a43b/ : SUCCESS in 25m 43s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/68a6b68/ : SUCCESS in 16m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-unit-latest-ubuntu-bionic/7d791ea/ : FAILURE in 20m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/14/659414/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/31d430a/ : SUCCESS in 58m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/14/659414/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/957aca5/ : SUCCESS in 1h 02m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/14/659414/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/1c66018/ : SUCCESS in 1h 00m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/14/659414/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/b60c74c/ : SUCCESS in 47m 20s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/14/659414/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/aa74dd1/ : FAILURE in 31m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/94efa7b/ : FAILURE in 28m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/88998d2/ : FAILURE in 26m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/14/659414/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/0a32fd3/ : FAILURE in 21m 54s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/14/659414/1/check/puppet-openstack-beaker-centos-7/ada83dc/ : SUCCESS in 17m 31s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/14/659414/1/check/puppet-openstack-beaker-ubuntu-bionic/bb8ad75/ : FAILURE in 10m 48s (non-voting)\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/14/659414/1/check/puppet-neutron-tripleo-standalone/582533d/ : SUCCESS in 3h 05m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75f02900704e3178843273ed16926b06da8d2627","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 01:01:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69bb736bde180a6bcba8ccd93fdb713fb780db23","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 02:41:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/14/659414/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/e7ec0de/ : FAILURE in 1h 30m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/14/659414/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/54dc65f/ : SUCCESS in 1h 34m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/14/659414/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/2b05007/ : SUCCESS in 1h 36m 45s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/14/659414/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/66d91c0/ : SUCCESS in 1h 08m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76656e29d5c8fc89c51de6f63d083487ca510339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 04:08:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-lint-ubuntu-bionic/a16b252/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-syntax-4-ubuntu-bionic/7f97b99/ : SUCCESS in 3m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-syntax-5-ubuntu-bionic/00ace1a/ : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-syntax-6-ubuntu-bionic/f332e64/ : SUCCESS in 3m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/14/659414/2/check/puppet-openstack-unit-4.8-centos-7/922fc33/ : SUCCESS in 35m 34s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/14/659414/2/check/puppet-openstack-unit-5.5-centos-7/a76886e/ : SUCCESS in 22m 30s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/9d34020/ : SUCCESS in 14m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-unit-latest-ubuntu-bionic/356d7a4/ : FAILURE in 13m 36s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/14/659414/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/bb93524/ : SUCCESS in 49m 31s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/14/659414/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/eb3ee4b/ : SUCCESS in 57m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/14/659414/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/98c27aa/ : SUCCESS in 1h 00m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/14/659414/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/316895f/ : SUCCESS in 45m 35s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/14/659414/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/e52cae7/ : FAILURE in 28m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/97d75a7/ : FAILURE in 30m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/b8ca46e/ : FAILURE in 22m 22s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/14/659414/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/84e6f9b/ : FAILURE in 22m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/14/659414/2/check/puppet-openstack-beaker-centos-7/24d3ad2/ : SUCCESS in 16m 41s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/14/659414/2/check/puppet-openstack-beaker-ubuntu-bionic/29f86f8/ : FAILURE in 9m 03s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/14/659414/2/check/build-openstack-releasenotes/5829f08/html/ : SUCCESS in 5m 17s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/14/659414/2/check/puppet-neutron-tripleo-standalone/be1c492/ : SUCCESS in 2h 54m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ae0af10639c5073bd88bea30e393544197f2ffd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 08:32:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23f3eb7e2dd5bdde633a705255431128abea6688","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 10:41:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/14/659414/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/1cf882d/ : SUCCESS in 1h 14m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/14/659414/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/1652164/ : SUCCESS in 1h 15m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/14/659414/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/6cfa0a3/ : SUCCESS in 1h 23m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/14/659414/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/2ab63ec/ : SUCCESS in 58m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"198e648e3ea7b79030cba132ccc727dcd681e5ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 12:05:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-lint-ubuntu-bionic/57876fb/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-syntax-4-ubuntu-bionic/981f175/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-syntax-5-ubuntu-bionic/f8641d8/ : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-syntax-6-ubuntu-bionic/3bc54b0/ : SUCCESS in 3m 49s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-unit-4.8-centos-7/a6cf973/ : SUCCESS in 40m 26s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-unit-5.5-centos-7/5848b0f/ : SUCCESS in 31m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/da97eab/ : SUCCESS in 13m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-unit-latest-ubuntu-bionic/0194502/ : FAILURE in 17m 29s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/e969f86/ : SUCCESS in 57m 44s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/bcb9ed8/ : SUCCESS in 59m 45s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/ae8b799/ : SUCCESS in 56m 43s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/b2fd800/ : SUCCESS in 52m 29s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/ba51eaa/ : FAILURE in 27m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/e8d33bd/ : FAILURE in 31m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/3e07395/ : FAILURE in 37m 53s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/c65e246/ : FAILURE in 21m 34s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-beaker-centos-7/7c41bb9/ : SUCCESS in 17m 18s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-beaker-ubuntu-bionic/86d5a6e/ : FAILURE in 8m 08s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/14/659414/3/check/build-openstack-releasenotes/a9b5d5d/html/ : SUCCESS in 3m 27s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/14/659414/3/check/puppet-neutron-tripleo-standalone/df4188d/ : FAILURE in 3h 09m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a1499b4a0de52ddd9931ecebcbf9d447813555d3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-17 19:27:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5b1fe3ca74cc0881da2763c66d3ef9c4668d0ecd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-17 19:27:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fca873d8979ee780dd290e0901f4a8114e451c87","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 01:04:54.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1b7a434c6ddcf6fed2a669d8efa1036781e4b895","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 01:05:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dc759696dd0db2edacd3eac54bc8de60d31a7186","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:44:04.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-lint-ubuntu-bionic/f48f49c/ : SUCCESS in 3m 57s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-syntax-4-ubuntu-bionic/8d67e85/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/92bed47/ : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/c193018/ : SUCCESS in 3m 26s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-unit-4.8-centos-7/9a0f254/ : SUCCESS in 27m 50s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-unit-5.5-centos-7/fd027ad/ : SUCCESS in 23m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/a210302/ : SUCCESS in 11m 33s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/14/659414/3/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/26b0165/ : SUCCESS in 58m 57s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/1756fdf/ : SUCCESS in 1h 01m 39s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/75288fe/ : SUCCESS in 54m 44s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/14/659414/3/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/d802c28/ : FAILURE in 53m 06s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-beaker-centos-7/7c77d3c/ : SUCCESS in 16m 57s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-beaker-ubuntu-bionic/f01d263/ : FAILURE in 8m 07s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/14/659414/3/gate/build-openstack-releasenotes/5ef6cec/html/ : SUCCESS in 2m 03s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/14/659414/3/gate/puppet-neutron-tripleo-standalone/89c6487/ : SUCCESS in 2h 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ead8d99d3223f09277c0b67dddb7f83dc75579a1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 03:50:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"db5a1ac25112e47ca17bac8100b70a3da39001b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 04:49:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/14/659414/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/db15b48/ : SUCCESS in 51m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/14/659414/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/4bd4b4a/ : SUCCESS in 56m 49s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/14/659414/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/25d1146/ : SUCCESS in 54m 11s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/14/659414/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/1886850/ : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a02d8b197faf13a9cfb2c7a9a6bce512a84f0afe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 06:35:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-lint-ubuntu-bionic/3cf12c3/ : SUCCESS in 4m 44s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-syntax-4-ubuntu-bionic/d3ba183/ : SUCCESS in 4m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-syntax-5-ubuntu-bionic/5a58e48/ : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-syntax-6-ubuntu-bionic/216cf73/ : SUCCESS in 3m 37s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-unit-4.8-centos-7/8139d7a/ : SUCCESS in 26m 45s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-unit-5.5-centos-7/eb43b08/ : SUCCESS in 26m 57s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/9d8e9ea/ : SUCCESS in 16m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-unit-latest-ubuntu-bionic/235be53/ : FAILURE in 15m 14s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/a42fc07/ : SUCCESS in 50m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/578bf97/ : SUCCESS in 1h 03m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/5d5cc47/ : SUCCESS in 55m 29s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/76b0632/ : SUCCESS in 49m 52s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/5780708/ : FAILURE in 29m 24s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/eafd28b/ : FAILURE in 40m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/bc49725/ : FAILURE in 29m 17s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/14/659414/3/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/f80416a/ : FAILURE in 22m 07s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/14/659414/3/check/puppet-openstack-beaker-centos-7/f414fc6/ : SUCCESS in 25m 56s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/14/659414/3/check/puppet-openstack-beaker-ubuntu-bionic/10e8625/ : FAILURE in 9m 36s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/14/659414/3/check/build-openstack-releasenotes/2203399/html/ : SUCCESS in 2m 30s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/14/659414/3/check/puppet-neutron-tripleo-standalone/8500b4e/ : SUCCESS in 2h 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ecb31ebc24ca809027a0a3e4c1c988d277f85fbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 06:36:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01d069231ad405c2a0f4cc0e6686e98598530730","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:39:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"33fbf340d067945771223a1b93f379ab9bd6741e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:39:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-lint-ubuntu-bionic/a39bd9b/ : SUCCESS in 4m 32s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-syntax-4-ubuntu-bionic/0fd2061/ : SUCCESS in 4m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/a84ceb8/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/23dcf84/ : SUCCESS in 4m 07s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-unit-4.8-centos-7/82a737c/ : SUCCESS in 29m 20s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-unit-5.5-centos-7/1d5e712/ : SUCCESS in 23m 21s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/8fc7820/ : SUCCESS in 16m 59s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/14/659414/3/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/7a53e65/ : SUCCESS in 57m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/692c845/ : SUCCESS in 59m 03s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/ce13659/ : SUCCESS in 1h 02m 10s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/14/659414/3/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/2e13fc8/ : SUCCESS in 46m 34s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/14/659414/3/gate/puppet-openstack-beaker-centos-7/0be48a1/ : SUCCESS in 25m 06s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/14/659414/3/gate/puppet-openstack-beaker-ubuntu-bionic/05dc5f1/ : FAILURE in 7m 31s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/14/659414/3/gate/build-openstack-releasenotes/d8fc5f7/html/ : SUCCESS in 2m 49s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/14/659414/3/gate/puppet-neutron-tripleo-standalone/a91c690/ : SUCCESS in 3h 03m 05s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"33574debba79de0c29b25a69c620fe288ef60b75","revisions":{"cc97b39ee79c345306d897a09a2617936df18e0c":{"kind":"REWORK","_number":1,"created":"2019-05-16 01:53:16.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/14/659414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/14/659414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/1"}}},"commit":{"parents":[{"commit":"dd3ba5ed0a624ac862b257642b6d47ddc32e6d14","subject":"Merge \"Unified agent section of ini files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dd3ba5ed0a624ac862b257642b6d47ddc32e6d14"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:25.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:25.000000000","tz":480},"subject":"Add openstackclient","message":"Add openstackclient\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I857de092429e94890064a16475c0442f3751ae45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cc97b39ee79c345306d897a09a2617936df18e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/cc97b39ee79c345306d897a09a2617936df18e0c"}]},"branch":"refs/heads/master"},"793ea676a7f53508d5bdcae75a1a9b6d0d78fdb8":{"kind":"REWORK","_number":2,"created":"2019-05-17 01:01:32.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/14/659414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/14/659414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/2"}}},"commit":{"parents":[{"commit":"dd3ba5ed0a624ac862b257642b6d47ddc32e6d14","subject":"Merge \"Unified agent section of ini files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dd3ba5ed0a624ac862b257642b6d47ddc32e6d14"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:25.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 01:00:25.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I857de092429e94890064a16475c0442f3751ae45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/793ea676a7f53508d5bdcae75a1a9b6d0d78fdb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/793ea676a7f53508d5bdcae75a1a9b6d0d78fdb8"}]},"branch":"refs/heads/master"},"33574debba79de0c29b25a69c620fe288ef60b75":{"kind":"REWORK","_number":3,"created":"2019-05-17 08:32:54.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/14/659414/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/14/659414/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/14/659414/3"}}},"commit":{"parents":[{"commit":"dd3ba5ed0a624ac862b257642b6d47ddc32e6d14","subject":"Merge \"Unified agent section of ini files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dd3ba5ed0a624ac862b257642b6d47ddc32e6d14"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:25.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 08:31:19.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I857de092429e94890064a16475c0442f3751ae45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/33574debba79de0c29b25a69c620fe288ef60b75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/33574debba79de0c29b25a69c620fe288ef60b75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
