)]}'
{"id":"openstack%2Fpuppet-neutron~664419","triplet_id":"openstack%2Fpuppet-neutron~master~I8248ad80e6ff88aec252919fa687018bda78a60b","project":"openstack/puppet-neutron","branch":"master","topic":"bug/1832251","hashtags":[],"change_id":"I8248ad80e6ff88aec252919fa687018bda78a60b","subject":"Ensure tunnel types are absent if not configured","status":"MERGED","created":"2019-06-10 18:26:01.000000000","updated":"2019-06-13 19:37:20.000000000","submitted":"2019-06-13 19:37:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664419-1560454640426-216c3fa0","meta_rev_id":"97ccda1592b175b22a48a4793d8918876cd3901f","_number":664419,"virtual_id_number":664419,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-13 19:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-10 19:14:18.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-06-12 00:29:59.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-06-13 14:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-06-13 14:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 19:14:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 00:29:59.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-13 14:59:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-13 19:37:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"580decb0a667ac672580635fee9b8a0686a83745","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-10 18:26:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f72f6d741be0512bc28cf3a88ef145bfd4cd782c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 19:14:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/19/664419/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/058d847/ : SUCCESS in 33m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/19/664419/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/4b684f0/ : SUCCESS in 46m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/19/664419/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/bc87d7c/ : SUCCESS in 37m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/19/664419/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/5850b06/ : SUCCESS in 29m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5c9e95a29789efdc671c079412718a2790e2321","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:48:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-lint-ubuntu-bionic/327cb8d/ : SUCCESS in 5m 42s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-syntax-4-ubuntu-bionic/4e18bf9/ : SUCCESS in 4m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-syntax-5-ubuntu-bionic/5e8c6ee/ : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-syntax-6-ubuntu-bionic/8e37f17/ : SUCCESS in 6m 40s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/664419/1/check/puppet-openstack-unit-4.8-centos-7/951c8b7/ : SUCCESS in 31m 58s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/19/664419/1/check/puppet-openstack-unit-5.5-centos-7/0b1f5d5/ : SUCCESS in 28m 06s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/d06aede/ : SUCCESS in 16m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-unit-latest-ubuntu-bionic/fef8af9/ : FAILURE in 20m 18s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/19/664419/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/828920d/ : SUCCESS in 1h 00m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/19/664419/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/80760b9/ : SUCCESS in 1h 10m 35s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/19/664419/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/abec1d8/ : SUCCESS in 1h 06m 43s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/19/664419/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/b7f5f92/ : SUCCESS in 47m 19s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/19/664419/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/cbcf918/ : FAILURE in 3m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/baad182/ : FAILURE in 32m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/c95bc30/ : FAILURE in 28m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/19/664419/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/cc0e2ef/ : FAILURE in 27m 06s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/19/664419/1/check/puppet-openstack-beaker-centos-7/8011000/ : SUCCESS in 39m 02s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/19/664419/1/check/puppet-openstack-beaker-ubuntu-bionic/7f97955/ : FAILURE in 10m 00s (non-voting)\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/19/664419/1/check/puppet-neutron-tripleo-standalone/3d63d70/ : FAILURE in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a58d7b5b807e693f9211ff967a75c59d9eb6bdb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 00:29:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ebdc39f3f2658145464fca8b73b04789acc05bdd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-13 14:59:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a261ab5d5f00bd0a7f2ea45eb5f9d47864aaede8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 15:04:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a53e8153418f202d08dc72f94d7bdfd1c179aa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 16:01:04.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebb064bb33e24f70b3202325d0f8c2e56e505f12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 19:37:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"131232bfb789178480be8c886456c1d2e540a7c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 19:37:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/19/664419/1/gate/puppet-openstack-lint-ubuntu-bionic/33bd84b/ : SUCCESS in 7m 32s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/19/664419/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/d29a5e8/ : SUCCESS in 8m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/19/664419/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/d558d9b/ : SUCCESS in 8m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/19/664419/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/8a7e95d/ : SUCCESS in 8m 12s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/664419/1/gate/puppet-openstack-unit-4.8-centos-7/ee1b278/ : SUCCESS in 36m 20s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/19/664419/1/gate/puppet-openstack-unit-5.5-centos-7/1f0f08e/ : SUCCESS in 34m 25s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/19/664419/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/c8df95e/ : SUCCESS in 17m 39s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/19/664419/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/a520359/ : SUCCESS in 52m 30s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/19/664419/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/8268372/ : SUCCESS in 1h 03m 49s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/19/664419/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/f92514a/ : SUCCESS in 1h 10m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/19/664419/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/ac26243/ : SUCCESS in 46m 11s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/19/664419/1/gate/puppet-openstack-beaker-centos-7/0b12690/ : SUCCESS in 21m 37s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/19/664419/1/gate/puppet-openstack-beaker-ubuntu-bionic/dc1db70/ : FAILURE in 10m 35s (non-voting)\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/19/664419/1/gate/puppet-neutron-tripleo-standalone/c979191/ : SUCCESS in 3h 23m 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c34ce2a42ae8c445a1fbe14ab18d4ee59e09c9b","revisions":{"6c34ce2a42ae8c445a1fbe14ab18d4ee59e09c9b":{"kind":"REWORK","_number":1,"created":"2019-06-10 18:26:01.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/19/664419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/19/664419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/19/664419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/19/664419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/19/664419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/19/664419/1"}}},"commit":{"parents":[{"commit":"d63c34681bc0b133f7460ac7dd21a17973fb3353","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d63c34681bc0b133f7460ac7dd21a17973fb3353"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-10 18:24:23.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-10 18:26:00.000000000","tz":-150},"subject":"Ensure tunnel types are absent if not configured","message":"Ensure tunnel types are absent if not configured\n\nThis patch fixes a bug where updating a configuration wouldn\u0027t remove\nagent/tunnel_types configuration when required.\n\nChange-Id: I8248ad80e6ff88aec252919fa687018bda78a60b\nCloses-Bug: #1832251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6c34ce2a42ae8c445a1fbe14ab18d4ee59e09c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6c34ce2a42ae8c445a1fbe14ab18d4ee59e09c9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
