)]}'
{"id":"openstack%2Fpuppet-neutron~665801","triplet_id":"openstack%2Fpuppet-neutron~master~I799a9e49fb2e5310c9489a06a520d2c2576c2637","project":"openstack/puppet-neutron","branch":"master","topic":"665801","hashtags":[],"change_id":"I799a9e49fb2e5310c9489a06a520d2c2576c2637","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:25:59.000000000","updated":"2020-09-18 22:56:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c288a6c3d30446493a001ee160b47fe49a4ca9b5","_number":665801,"virtual_id_number":665801,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-18 02:43:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-18 01:00:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 01:00:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 02:43:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ae5b21231dab38e671d7468db4a23bc0653f4bf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:25:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81cb6e3906f28b50055bcdc989cd27aaca1fe9af","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:48:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed1fefc433f5b2b82710a37261d1f1f5b03d7c00","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 01:00:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/01/665801/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/ab73d7b/ : SUCCESS in 37m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/01/665801/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/2104ec8/ : SUCCESS in 43m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/01/665801/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/a03daf2/ : SUCCESS in 1h 04m 04s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/01/665801/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/73dfee9/ : SUCCESS in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f17aa7250ac5e002824300e5f3377b9f2d60c0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 02:43:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/01/665801/2/check/puppet-openstack-lint-ubuntu-bionic/b060fc0/ : SUCCESS in 4m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/01/665801/2/check/puppet-openstack-syntax-5-ubuntu-bionic/cf673a7/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/01/665801/2/check/puppet-openstack-syntax-6-ubuntu-bionic/88d903d/ : SUCCESS in 4m 52s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/665801/2/check/puppet-openstack-unit-5.5-centos-7/e60b587/ : FAILURE in 24m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/01/665801/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/3a5d971/ : FAILURE in 13m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/01/665801/2/check/puppet-openstack-unit-latest-ubuntu-bionic/c421fee/ : FAILURE in 17m 41s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/01/665801/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/634bf88/ : SUCCESS in 52m 44s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/01/665801/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/42f4539/ : SUCCESS in 1h 02m 52s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/01/665801/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/6dfbfcc/ : SUCCESS in 1h 03m 15s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/01/665801/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/1eaa458/ : SUCCESS in 41m 57s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/01/665801/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/6f078d5/ : FAILURE in 4m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/01/665801/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/da06cec/ : FAILURE in 31m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/01/665801/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/816fbfe/ : FAILURE in 25m 22s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/01/665801/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/2f0f5ee/ : FAILURE in 25m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/01/665801/2/check/puppet-openstack-beaker-centos-7/e20a883/ : SUCCESS in 18m 16s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/01/665801/2/check/puppet-openstack-beaker-ubuntu-bionic/7f87bf8/ : FAILURE in 9m 11s (non-voting)\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/01/665801/2/check/puppet-neutron-tripleo-standalone/8820d8a/ : SUCCESS in 2h 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74c5f1e4dc0776a77a2cd3919fb4d291f56553d1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:56:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0305ed7331c6550cd5166cd9f3c1d8f496f2e95f","revisions":{"bcb307771175a9ac05d50ac4bc11f741e748d9e0":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:25:59.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/01/665801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/665801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/665801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/665801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/665801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/665801/1"}}},"commit":{"parents":[{"commit":"b5769833436962a76f46ebe181859dc6db3973a3","subject":"Merge \"Ensure tunnel types are absent if not configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b5769833436962a76f46ebe181859dc6db3973a3"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:25:50.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:25:50.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I799a9e49fb2e5310c9489a06a520d2c2576c2637\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bcb307771175a9ac05d50ac4bc11f741e748d9e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bcb307771175a9ac05d50ac4bc11f741e748d9e0"}]},"branch":"refs/heads/master"},"0305ed7331c6550cd5166cd9f3c1d8f496f2e95f":{"kind":"REWORK","_number":2,"created":"2019-06-17 23:48:59.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/01/665801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/01/665801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/665801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/665801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/01/665801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/01/665801/2"}}},"commit":{"parents":[{"commit":"b5769833436962a76f46ebe181859dc6db3973a3","subject":"Merge \"Ensure tunnel types are absent if not configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b5769833436962a76f46ebe181859dc6db3973a3"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:25:50.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:48:53.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I799a9e49fb2e5310c9489a06a520d2c2576c2637\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0305ed7331c6550cd5166cd9f3c1d8f496f2e95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0305ed7331c6550cd5166cd9f3c1d8f496f2e95f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
